Skip to content

Accessibility engineering review for Link component. - #3317

Merged
radglob merged 44 commits into
mainfrom
link-a11y-review
Jun 22, 2023
Merged

Accessibility engineering review for Link component.#3317
radglob merged 44 commits into
mainfrom
link-a11y-review

Conversation

@radglob

@radglobradglob commented May 23, 2023

Copy link
Copy Markdown
Contributor

Restricting the as prop in Link components. Leaving forcing underlining on links in text to more holistic changes.

Closeshttps://github.com/github/primer/issues/1881

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@radglobradglob added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label May 23, 2023
@radglob
radglob requested review from a team and siddharthkpMay 23, 2023 20:29
@changeset-bot

changeset-botBot commented May 23, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4aba402

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented May 23, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js101.14 KB (+0.13% 🔺)
dist/browser.umd.js101.71 KB (+0.14% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 23, 2023 20:35 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 23, 2023 20:37 Inactive
@radglob
radglob temporarily deployed to github-pages May 23, 2023 20:40 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 23, 2023 20:40 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 25, 2023 15:25 Inactive
@radglob
radglob temporarily deployed to github-pages May 25, 2023 15:29 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 25, 2023 15:29 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 25, 2023 15:30 Inactive
@radglob
radglob temporarily deployed to github-pages May 25, 2023 19:44 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 25, 2023 19:45 Inactive
@radglob
radglob temporarily deployed to github-pages May 26, 2023 13:30 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 26, 2023 13:30 Inactive
@radglob
radglob temporarily deployed to github-pages May 26, 2023 16:43 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 May 26, 2023 16:44 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 June 8, 2023 13:18 Inactive
@radglob
radglob temporarily deployed to github-pages June 8, 2023 20:28 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 June 8, 2023 20:29 Inactive
@lesliecdubs

Copy link
Copy Markdown
Member

@TylerJDev could you help out with a review on this one? Thanks!

@TylerJDevTylerJDev left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good to me! I think there's some pending discussion that'll need to be addressed first, but other than that I think that this is great!

@radglob
radglob temporarily deployed to github-pages June 15, 2023 20:50 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 June 15, 2023 20:50 Inactive
@radglob
radglob temporarily deployed to github-pages June 16, 2023 16:23 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 June 16, 2023 16:23 Inactive
@radglob
radglob temporarily deployed to github-pages June 16, 2023 17:56 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 June 16, 2023 17:56 Inactive
@radglob
radglob temporarily deployed to github-pages June 16, 2023 18:09 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 June 16, 2023 18:10 Inactive
@radglob
radglob temporarily deployed to github-pages June 20, 2023 13:25 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3317 June 20, 2023 13:25 Inactive
Comment threadsrc/Button/types.ts
Comment threadsrc/Button/styles.ts Outdated
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

update snapshots🤖 Command that updates VRT snapshots on the pull request

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants

@radglob@lesliecdubs@mperrotti@langermank@TylerJDev