Skip to content

SelectPanel: Add subtitle prop - #3322

Merged
colebemis merged 7 commits into
mainfrom
selectpanel-subtitle-prop
Jun 7, 2023
Merged

SelectPanel: Add subtitle prop#3322
colebemis merged 7 commits into
mainfrom
selectpanel-subtitle-prop

Conversation

@colebemis

@colebemiscolebemis commented May 24, 2023

Copy link
Copy Markdown
Contributor

Adds an optional subtitle prop to SelectPanel:

Part of https://github.com/github/primer/issues/2087

Screenshots

CleanShot 2023-05-24 at 12 47 40@2x

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented May 24, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f01e39a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented May 24, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js101.86 KB (+0.04% 🔺)
dist/browser.umd.js102.35 KB (+0.05% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3322 May 24, 2023 19:04 Inactive
@primer
primerBot temporarily deployed to github-pages May 24, 2023 19:05 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3322 May 24, 2023 19:05 Inactive
@colebemis
colebemis marked this pull request as ready for review May 24, 2023 19:53
@colebemis
colebemis requested review from a team and joshblackMay 24, 2023 19:53
@colebemis
colebemis temporarily deployed to github-pages May 24, 2023 19:59 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3322 May 24, 2023 19:59 Inactive
@colebemis
colebemis temporarily deployed to github-pages May 24, 2023 20:23 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3322 May 24, 2023 20:23 Inactive
@colebemiscolebemis added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label May 24, 2023
@colebemiscolebemis added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Jun 7, 2023
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jun 7, 2023
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3322 June 7, 2023 18:35 Inactive
@colebemis
colebemis added this pull request to the merge queueJun 7, 2023
Merged via the queue into main with commit c0b2fcfJun 7, 2023
@colebemis
colebemis deleted the selectpanel-subtitle-prop branch June 7, 2023 19:05
This was referenced Jun 7, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@colebemis@joshblack