Skip to content

Upgrade styled-components to v5 - #3397

Merged
mperrotti merged 17 commits into
mainfrom
mp/styled-components-v5-upgrade
Jul 26, 2023
Merged

Upgrade styled-components to v5#3397
mperrotti merged 17 commits into
mainfrom
mp/styled-components-v5-upgrade

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Jun 9, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c0dac98

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment on lines +52 to +62
@media screen and (min-width:544px) {

}

@media screen and (min-width:768px) {

}

@media screen and (min-width:1012px) {

}

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Not sure what's causing these to get output

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Seems like this is a know issue both for us and for the community

That said since we are moving away from jest snapshots (ADR reference), if this is only snapshot issue (seems like it) we might maybe want to start deleting these snapshots? At least we can start with Box maybe? And if needed we can create a test in Playwright. What do you think?

@mperrotti
mperrotti temporarily deployed to github-pages June 9, 2023 15:34 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 June 9, 2023 15:35 Inactive
@github-actions

github-actionsBot commented Jun 9, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js102.48 KB (0%)
dist/browser.umd.js103.08 KB (0%)

@mperrotti
mperrotti temporarily deployed to github-pages June 9, 2023 16:27 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 June 9, 2023 16:27 Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great to me 🥳 What would the best next steps for this be? Would it be going through and doing a visual inspection to make sure nothing was impacted?

Comment thread.changeset/proud-rules-develop.md Outdated
mperrottiand others added 2 commits July 7, 2023 11:37
@mperrotti
mperrotti temporarily deployed to github-pages July 7, 2023 15:45 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 July 7, 2023 15:45 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 July 7, 2023 15:46 Inactive
@mperrotti

Copy link
Copy Markdown
ContributorAuthor

@joshblack - I already went through a bunch of stuff in Storybook, but maybe we could visually review in dotcom

@mperrotti
mperrotti temporarily deployed to github-pages July 11, 2023 14:48 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 July 11, 2023 14:48 Inactive
@mperrotti
mperrotti marked this pull request as ready for review July 11, 2023 23:09
@mperrotti
mperrotti requested review from a team and broccolinisoupJuly 11, 2023 23:09
@mperrotti
mperrotti temporarily deployed to github-pages July 11, 2023 23:14 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 July 11, 2023 23:15 Inactive
@mperrotti
mperrotti temporarily deployed to github-pages July 12, 2023 20:43 — with GitHub Actions Inactive
@mperrotti

Copy link
Copy Markdown
ContributorAuthor

if this is only snapshot issue (seems like it) we might maybe want to start deleting these snapshots? At least we can start with Box maybe? And if needed we can create a test in Playwright. What do you think?

@broccolinisoup - I think we can just delete that snapshot

@mperrotti
mperrotti temporarily deployed to github-pages July 19, 2023 16:02 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 July 19, 2023 16:02 Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🥳

@mperrotti
mperrotti temporarily deployed to github-pages July 20, 2023 13:35 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 July 20, 2023 13:35 Inactive
@broccolinisoup

Copy link
Copy Markdown
Member

if this is only snapshot issue (seems like it) we might maybe want to start deleting these snapshots? At least we can start with Box maybe? And if needed we can create a test in Playwright. What do you think?

@broccolinisoup - I think we can just delete that snapshot

I think so too!

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🥳

@mperrotti
mperrotti temporarily deployed to github-pages July 25, 2023 17:35 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3397 July 25, 2023 17:35 Inactive
@mperrotti
mperrotti enabled auto-merge July 26, 2023 23:15
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3397 July 26, 2023 23:22 Inactive
@mperrotti
mperrotti added this pull request to the merge queueJul 26, 2023
Merged via the queue into main with commit 397938dJul 26, 2023
@mperrotti
mperrotti deleted the mp/styled-components-v5-upgrade branch July 26, 2023 23:39
@primer-cssprimer-css mentioned this pull request Jul 26, 2023
joshblack added a commit that referenced this pull request Jul 27, 2023
github-merge-queueBot pushed a commit that referenced this pull request Jul 27, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@mperrotti@broccolinisoup@joshblack