Skip to content

🛠️ [Bug Fix] Setting DialogV2 backdrop to the proper variable - #3411

Merged
joshblack merged 2 commits into
mainfrom
dusave/dialogv2-backdrop-fix
Jun 20, 2023
Merged

🛠️ [Bug Fix] Setting DialogV2 backdrop to the proper variable#3411
joshblack merged 2 commits into
mainfrom
dusave/dialogv2-backdrop-fix

Conversation

@dusave

Copy link
Copy Markdown
Contributor

This fixes a bug where the DialogV2 backdrop is not set to the proper CSS variable

Screenshots

Before:
image
background-color: rgba(0, 0, 0, 0.4);

After:
image
background-color: rgba(31,35,40,0.5);

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@dusave
dusave requested review from a team and colebemisJune 13, 2023 20:09
@changeset-bot

changeset-botBot commented Jun 13, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 78253d0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

Uh oh! @dusave, the image you shared is missing helpful alt text. Check your pull request body.

Alt text is an invisible description that helps screen readers describe images to blind or low-vision users. If you are using markdown to display images, add your alt text inside the brackets of the markdown image.

Learn more about alt text at Basic writing and formatting syntax: images on GitHub Docs.

@github-actions

github-actionsBot commented Jun 13, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js101.87 KB (0%)
dist/browser.umd.js102.37 KB (+0.01% 🔺)

@dusave
dusave temporarily deployed to github-pages June 13, 2023 20:15 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3411 June 13, 2023 20:15 Inactive

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thank you!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-3411 June 13, 2023 20:29 Inactive
@dusave
dusave enabled auto-merge June 13, 2023 20:47
@dusave
dusave disabled auto-merge June 13, 2023 21:00
@joshblack
joshblack added this pull request to the merge queueJun 19, 2023
@joshblack
joshblack removed this pull request from the merge queue due to a manual request Jun 19, 2023
@joshblack

Copy link
Copy Markdown
Member

@dusave was adding this to the merge queue but saw that you disabled auto merge last week (sorry about that!) Let me know if this is good to go or if you wanted to make any additional changes!

@dusave

Copy link
Copy Markdown
ContributorAuthor

@joshblack ah, thanks for the heads up! I was meaning to follow up on the merge process.

This bad boy is ready to go!

@joshblack
joshblack added this pull request to the merge queueJun 20, 2023
Merged via the queue into main with commit 295c4bcJun 20, 2023
@joshblack
joshblack deleted the dusave/dialogv2-backdrop-fix branch June 20, 2023 15:11
This was referenced Jun 20, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@dusave@joshblack@colebemis