Skip to content

[WIP] Fix misc issues with Dialog v2 - #3443

Closed
mperrotti wants to merge 5 commits into
next-majorfrom
mp/dialogV2-doc-and-style-fixes
Closed

[WIP] Fix misc issues with Dialog v2#3443
mperrotti wants to merge 5 commits into
next-majorfrom
mp/dialogV2-doc-and-style-fixes

Conversation

@mperrotti

@mperrottimperrotti commented Jun 22, 2023

Copy link
Copy Markdown
Contributor

Closes # https://github.com/github/primer/issues/2372

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

joshblackand others added 3 commits June 21, 2023 17:36
* fix: add experimental folder to npm files
* chore: add changeset
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
@changeset-bot

changeset-botBot commented Jun 22, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e802b24

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mperrottimperrotti changed the title Fix misc issues with Dialog v2[WIP] Fix misc issues with Dialog v2Jun 22, 2023
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3443 June 22, 2023 16:51 Inactive
@mperrotti
mperrotti temporarily deployed to github-pages June 22, 2023 16:52 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3443 June 22, 2023 16:52 Inactive
@mperrotti
mperrotti changed the base branch from main to next-majorJune 22, 2023 18:24
@mperrotti

Copy link
Copy Markdown
ContributorAuthor

Closing this and re-opening as a branch off of next-major

@mperrotti

Copy link
Copy Markdown
ContributorAuthor

New issue here: #3444

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@mperrotti@joshblack@broccolinisoup