Skip to content

[⏪ Reverted] Show underline on Link when hovered or focused. - #3586

Merged
radglob merged 4 commits into
mainfrom
link-show-underline-on-hover-focus
Aug 3, 2023
Merged

[⏪ Reverted] Show underline on Link when hovered or focused.#3586
radglob merged 4 commits into
mainfrom
link-show-underline-on-hover-focus

Conversation

@radglob

Copy link
Copy Markdown
Contributor

Resolveshttps://github.com/github/accessibility-audits/issues/4502 (and likely other issues).

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@radglob
radglob requested review from a team and joshblackAugust 2, 2023 15:40
@changeset-bot

changeset-botBot commented Aug 2, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0e3edf6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Aug 2, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js103.68 KB (+0.03% 🔺)
dist/browser.umd.js104.22 KB (+0.01% 🔺)

@radglob
radglob temporarily deployed to github-pages August 2, 2023 15:47 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3586 August 2, 2023 15:48 Inactive
@radglobradglob added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 2, 2023
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3586 August 2, 2023 15:48 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 2, 2023
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3586 August 2, 2023 16:01 Inactive
@lesliecdubs

Copy link
Copy Markdown
Member

👋🏻 Let's aim to get a Primer Design reviewer to give this a thorough review before merging, as it will have a noticeable impact on the visual styles across a wide area of the product. @langermank would you be willing to give this 👀 s since I know you've worked on Link before?

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I think this change makes sense! The only thing that stands out to me is that we list it as a common mistake in our interface guidelines. But we don't really follow those guidelines for links across GitHub anyhow, and I think this change is still an improvement?

@radglob
radglob added this pull request to the merge queueAug 3, 2023
Merged via the queue into main with commit fbcb3b5Aug 3, 2023
@radglob
radglob deleted the link-show-underline-on-hover-focus branch August 3, 2023 17:23
@primer-cssprimer-css mentioned this pull request Aug 3, 2023
siddharthkp added a commit that referenced this pull request Aug 4, 2023
github-merge-queueBot pushed a commit that referenced this pull request Aug 4, 2023
@siddharthkpsiddharthkp changed the title Show underline on Link when hovered or focused.[⏪ Reverted] Show underline on Link when hovered or focused.Aug 7, 2023
siddharthkp added a commit that referenced this pull request Aug 7, 2023
siddharthkp added a commit that referenced this pull request Aug 7, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@radglob@lesliecdubs@langermank