Skip to content

Revert #3389 (PageLayout.Pane: Undo deprecation for position prop) - #3589

Merged
siddharthkp merged 9 commits into
mainfrom
revert-3389
Aug 4, 2023
Merged

Revert #3389 (PageLayout.Pane: Undo deprecation for position prop)#3589
siddharthkp merged 9 commits into
mainfrom
revert-3389

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Aug 4, 2023

Copy link
Copy Markdown
Member
production (before)upgrade branch (after)
Sidebar in hyperlist is on the left Sidebar in hyperlist is on the leftSidebar in hyperlist is on the right Sidebar in hyperlist is on the right

We would need to bring this change back in a way that it allows new behaviour without breaking existing behaviour.

  • Sidenote: It seems like the same pull request number is also merged in next-major? Link to next-major release tracking That seems very fishy? Was this not supposed to be in main at all?

@siddharthkp
siddharthkp requested review from a team and broccolinisoupAugust 4, 2023 14:26
@changeset-bot

changeset-botBot commented Aug 4, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3c49ffd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkp
siddharthkp requested review from joshblack and removed request for broccolinisoupAugust 4, 2023 14:27
@siddharthkpsiddharthkp changed the title Revert #3389 (Deprecate position prop for PageLayout.Pane)Revert #3389 (PageLayout.Pane: Undo deprecation for position prop)Aug 4, 2023
@siddharthkp

siddharthkp commented Aug 4, 2023

Copy link
Copy Markdown
MemberAuthor

FYI @radglob@broccolinisoup, sorry! :(

@siddharthkpsiddharthkp self-assigned this Aug 4, 2023
@github-actions

github-actionsBot commented Aug 4, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js103.67 KB (-0.02% 🔽)
dist/browser.umd.js104.21 KB (-0.01% 🔽)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3589 August 4, 2023 14:33 Inactive
@siddharthkp
siddharthkp temporarily deployed to github-pages August 4, 2023 14:35 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3589 August 4, 2023 14:36 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3589 August 4, 2023 14:36 Inactive
@siddharthkp
siddharthkp temporarily deployed to github-pages August 4, 2023 14:43 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3589 August 4, 2023 14:44 Inactive
@siddharthkpsiddharthkp added update snapshots 🤖 Command that updates VRT snapshots on the pull request patch release bug fixes, docs, housekeeping labels Aug 4, 2023
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 4, 2023
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3589 August 4, 2023 15:06 Inactive
@siddharthkp
siddharthkp temporarily deployed to github-pages August 4, 2023 15:09 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3589 August 4, 2023 15:10 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3589 August 4, 2023 15:30 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3589 August 4, 2023 15:33 Inactive
@siddharthkp
siddharthkp added this pull request to the merge queueAug 4, 2023
Merged via the queue into main with commit d598a53Aug 4, 2023
@siddharthkp
siddharthkp deleted the revert-3389 branch August 4, 2023 16:05
@primer-cssprimer-css mentioned this pull request Aug 3, 2023
siddharthkp added a commit that referenced this pull request Aug 7, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

patch releasebug fixes, docs, housekeeping

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@langermank