Skip to content

fix(Button): make children prop optional - #3661

Merged
joshblack merged 6 commits into
mainfrom
fix/update-button-children-to-optional
Sep 7, 2023
Merged

fix(Button): make children prop optional#3661
joshblack merged 6 commits into
mainfrom
fix/update-button-children-to-optional

Conversation

@joshblack

@joshblackjoshblack commented Aug 23, 2023

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/2601

Update our base ButtonProps so that children is optional. This allows for the following migration pattern to be supported for v36:

// Before<ButtonleadingIcon={ChatIcon}><Button.Counter>{5}</Button.Counter></Button>// After<ButtonleadingAction={ChatIcon}count={5}/>

@joshblack
joshblack requested review from a team and mperrottiAugust 23, 2023 20:25
@changeset-bot

changeset-botBot commented Aug 23, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6f0270a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Aug 23, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js100.69 KB (0%)
dist/browser.umd.js101.27 KB (0%)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3661 August 23, 2023 20:31 Inactive
@joshblack
joshblack temporarily deployed to github-pages August 23, 2023 20:31 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3661 August 23, 2023 20:32 Inactive
@joshblack
joshblack temporarily deployed to github-pages September 6, 2023 20:25 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3661 September 6, 2023 20:26 Inactive

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🚀

@github-actions
github-actionsBottemporarily deployed to storybook-preview-3661 September 7, 2023 16:14 Inactive
@joshblack
joshblack added this pull request to the merge queueSep 7, 2023
Merged via the queue into main with commit 25693b0Sep 7, 2023
@joshblack
joshblack deleted the fix/update-button-children-to-optional branch September 7, 2023 16:24
This was referenced Sep 7, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@broccolinisoup@langermank