Skip to content

fix(ActionList): place id on item with role - #3691

Merged
joshblack merged 7 commits into
mainfrom
fix/update-action-list
Oct 10, 2023
Merged

fix(ActionList): place id on item with role#3691
joshblack merged 7 commits into
mainfrom
fix/update-action-list

Conversation

@joshblack

@joshblackjoshblack commented Aug 30, 2023

Copy link
Copy Markdown
Member

Context: https://github.slack.com/archives/C01L618AEP9/p1693422238328979

The id on ActionList.Item should be placed on the item with the corresponding ARIA role so that a consumer can use that value to manage attributes like aria-activedescendant appropriately.

Warning
This change could be considered a breaking change if we consider the id placement as something that should be stable across minor versions.

@joshblack
joshblack requested review from a team and mperrottiAugust 30, 2023 19:26
@changeset-bot

changeset-botBot commented Aug 30, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0a60663

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Aug 30, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.7 KB (+0.01% 🔺)
dist/browser.umd.js105.28 KB (+0.01% 🔺)

@joshblack
joshblack temporarily deployed to github-pages August 30, 2023 19:32 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3691 August 30, 2023 19:32 Inactive

@mperrottimperrotti left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

CI should pass after updating snapshots.

@joshblack
joshblack temporarily deployed to github-pages September 6, 2023 20:28 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3691 September 6, 2023 20:29 Inactive
@joshblack

Copy link
Copy Markdown
MemberAuthor

@dusave checking upstream, it doesn't seem like this change introduces a regression so it could be safe to land. Wanted to double-check with you to see how you felt about this change.

I'll also ask the team to see if we're okay with this kind of change in a current minor release 👀

@joshblack
joshblack temporarily deployed to github-pages September 7, 2023 16:10 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3691 September 7, 2023 16:10 Inactive

@dusavedusave left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🚀 Awesome!

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3691 September 13, 2023 15:30 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3691 October 10, 2023 17:56 Inactive
@joshblack
joshblack added this pull request to the merge queueOct 10, 2023
Merged via the queue into main with commit f4648b1Oct 10, 2023
@joshblack
joshblack deleted the fix/update-action-list branch October 10, 2023 18:06
This was referenced Oct 10, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@mperrotti@dusave