Skip to content

feat(npm): export components deprecated in v36 from @primer/react/deprecated - #3719

Merged
joshblack merged 2 commits into
mainfrom
feat/add-deprecated-components-for-v36
Sep 8, 2023
Merged

feat(npm): export components deprecated in v36 from @primer/react/deprecated#3719
joshblack merged 2 commits into
mainfrom
feat/add-deprecated-components-for-v36

Conversation

@joshblack

Copy link
Copy Markdown
Member

Part of: https://github.com/github/primer/issues/2628

Add components deprecated for v35 to deprecated entrypoint in v35 so that we can migrate teams before the v36 release.

@joshblack
joshblack requested review from a team and siddharthkpSeptember 7, 2023 17:31
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0018024

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js100.69 KB (0%)
dist/browser.umd.js101.27 KB (0%)

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good!

@joshblack
joshblack added this pull request to the merge queueSep 8, 2023
Merged via the queue into main with commit 500e529Sep 8, 2023
@joshblack
joshblack deleted the feat/add-deprecated-components-for-v36 branch September 8, 2023 17:08
This was referenced Sep 8, 2023
@siddharthkpsiddharthkp changed the title feat(npm): move deprecated components in v36 to deprecatedfeat(npm): copy deprecated components in v36 to deprecatedSep 20, 2023
@siddharthkpsiddharthkp changed the title feat(npm): copy deprecated components in v36 to deprecatedfeat(npm): export components deprecated in v36 from deprecated/indexSep 20, 2023
@siddharthkpsiddharthkp changed the title feat(npm): export components deprecated in v36 from deprecated/indexfeat(npm): export components deprecated in v36 from @primer/react/deprecatedSep 20, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@siddharthkp