Uh oh!
There was an error while loading. Please reload this page.
fix(Link): update underline={false} to disable underline - #3838
fix(Link): update underline={false} to disable underline#3838joshblack wants to merge 4 commits into
Conversation
🦋 Changeset detectedLatest commit: 4963ae7 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
size-limit report 📦
|
mperrotti
commented
Oct 17, 2023
I really like this approach. The only thing that feels a little strange is that we don't usually differentiate between an |
mperrotti
commented
Oct 17, 2023
I pushed a commit to update |
joshblack
commented
Nov 30, 2023
Superseded by: #3946 |
This is a follow-up to the latest release which included support for underline preference in the
Linkcomponent: #3720I noticed when trying to update code upstream that
underline={false}does not work with this preference enabled. This PR changes this behavior so thatunderline={false}is respected regardless of if the preference is enabled.I'm curious what folks feel about this change 🤔 I couldn't think of an alternative if someone wanted to disable underlines in a specific area (like for cards in a project).
Changelog
New
Changed
underline={false}branch in Link to settext-decorationto'none' if the value isfalseNo UnderlineLink storyNo underlineLink storyRemoved
Rollout strategy