Skip to content

refactor Dialog to use <dialog> internally - #3840

Closed
keithamus wants to merge 2 commits into
mainfrom
refactor-dialog-to-use-dialog-internally
Closed

refactor Dialog to use <dialog> internally#3840
keithamus wants to merge 2 commits into
mainfrom
refactor-dialog-to-use-dialog-internally

Conversation

@keithamus

@keithamuskeithamus commented Oct 18, 2023

Copy link
Copy Markdown
Contributor

Refs https://github.com/github/primer/issues/2531

This changes Dialog to use the <dialog> element internally, which allows us to delete a fair amount of code and should resolve some bugs and complexity we have with the current dialog.

Changelog

New

Changed

Dialog now uses native <dialog> under the hood.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan

Testing & Reviewing

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Added/updated previews (Storybook)
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Oct 18, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 51c8194

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@keithamus
keithamus temporarily deployed to github-pages October 18, 2023 11:52 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3840 October 18, 2023 11:53 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch 2 times, most recently from e90fb09 to 92a5e9cCompareOctober 23, 2023 14:06
@keithamus
keithamus temporarily deployed to github-pages October 23, 2023 14:12 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3840 October 23, 2023 14:12 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from 92a5e9c to 1f2d335CompareOctober 23, 2023 15:13
@keithamus
keithamus temporarily deployed to github-pages October 23, 2023 15:21 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3840 October 23, 2023 15:21 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from 1f2d335 to 4ec1f2bCompareOctober 24, 2023 09:36
@github-actions

github-actionsBot commented Oct 24, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.6 KB (+0.13% 🔺)
dist/browser.umd.js105.26 KB (+0.17% 🔺)

@keithamus
keithamus temporarily deployed to github-pages October 24, 2023 09:43 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3840 October 24, 2023 09:43 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from 4ec1f2b to b12480dCompareOctober 25, 2023 08:44
@keithamus
keithamus temporarily deployed to github-pages October 25, 2023 08:52 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3840 October 25, 2023 08:53 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from b12480d to 54820e7CompareOctober 27, 2023 09:53
@keithamus
keithamus temporarily deployed to github-pages October 27, 2023 10:00 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3840 October 27, 2023 10:01 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from 54820e7 to 2e21933CompareOctober 27, 2023 10:33
@keithamus
keithamus temporarily deployed to github-pages October 27, 2023 10:41 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3840 October 27, 2023 10:41 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from 2e21933 to fbe8372CompareOctober 27, 2023 11:06
@keithamus
keithamus temporarily deployed to github-pages October 27, 2023 11:13 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3840 October 27, 2023 11:13 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch 2 times, most recently from ca66c14 to 85cbf4aCompareNovember 2, 2023 16:37
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 November 2, 2023 16:45 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from eec3b55 to 8c1a4f4CompareDecember 7, 2023 12:18
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 7, 2023 12:21 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from 8c1a4f4 to 62db52aCompareDecember 7, 2023 12:22
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 7, 2023 12:23 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 7, 2023 12:27 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from 62db52a to 8e8afffCompareDecember 7, 2023 14:15
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 7, 2023 14:20 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 7, 2023 14:37 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 7, 2023 15:06 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 7, 2023 15:21 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch from 46179aa to a9e6040CompareDecember 8, 2023 14:27
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 8, 2023 14:29 Inactive
@keithamus
keithamusforce-pushed the refactor-dialog-to-use-dialog-internally branch 2 times, most recently from b5151ba to a5033c6CompareDecember 8, 2023 14:32
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 8, 2023 14:32 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3840 December 8, 2023 14:34 Inactive
@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

1 similar comment
@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@keithamus@siddharthkp