Skip to content

Show border when there's only one avatar in an AvatarStack - #3916

Merged
strackoverflow merged 3 commits into
mainfrom
avatarstack-border
Nov 8, 2023
Merged

Show border when there's only one avatar in an AvatarStack#3916
strackoverflow merged 3 commits into
mainfrom
avatarstack-border

Conversation

@strackoverflow

Copy link
Copy Markdown
Member

Closes#3716

This PR updates AvatarStack to show a border when there is only one avatar in the stack, making it consistent with the appearance of the Avatar component. This is especially obvious when using a high-contrast theme.

BeforeAfter
before (invisible border)after (visible border)

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan

Testing & Reviewing

Merge checklist

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@strackoverflowstrackoverflow self-assigned this Nov 7, 2023
@strackoverflow
strackoverflow requested review from a team and siddharthkpNovember 7, 2023 17:57
@changeset-bot

changeset-botBot commented Nov 7, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8b09422

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 7, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.08 KB (+0.03% 🔺)
dist/browser.umd.js104.64 KB (+0.03% 🔺)

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I like it! Good solution 👍

@strackoverflow
strackoverflow added this pull request to the merge queueNov 8, 2023
Merged via the queue into main with commit c2f81b3Nov 8, 2023
@strackoverflow
strackoverflow deleted the avatarstack-border branch November 8, 2023 17:00
@primerprimerBot mentioned this pull request Nov 8, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Avatar component has higher contrast than AvatarStack

2 participants

@strackoverflow@siddharthkp