Skip to content

Tooltip2: Make the text prop required - #3925

Merged
broccolinisoup merged 2 commits into
mainfrom
bs/tooltip-v2-text-required
Nov 9, 2023
Merged

Tooltip2: Make the text prop required #3925
broccolinisoup merged 2 commits into
mainfrom
bs/tooltip-v2-text-required

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Nov 9, 2023

Copy link
Copy Markdown
Member

Changelog

New

Changed

Update text prop from being optional to required.

Removed

Rollout strategy

This is technically a breaking change but because Tooltip is still experimental, breaking API changes are expected.

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan

Testing & Reviewing

Merge checklist

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@broccolinisoup
broccolinisoup requested review from a team and pksjceNovember 9, 2023 06:45
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ca3e552

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.05 KB (0%)
dist/browser.umd.js104.58 KB (0%)

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Makes perfect sense!

@broccolinisoup
broccolinisoup added this pull request to the merge queueNov 9, 2023
Merged via the queue into main with commit 6b362b2Nov 9, 2023
@broccolinisoup
broccolinisoup deleted the bs/tooltip-v2-text-required branch November 9, 2023 20:30
@primerprimerBot mentioned this pull request Nov 9, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@broccolinisoup@siddharthkp