Skip to content

Add 'unavailable' status to StatusLabel - #3963

Merged
mperrotti merged 6 commits into
mainfrom
mp/unavailable-label
Nov 28, 2023
Merged

Add 'unavailable' status to StatusLabel#3963
mperrotti merged 6 commits into
mainfrom
mp/unavailable-label

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Relates to https://github.com/github/primer/issues/2679

Screenshot 2023-11-21 at 2 48 44 PM

Changelog

New

The status prop now accepts 'unavailable'

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan

Testing & Reviewing

Merge checklist

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@mperrotti
mperrotti requested review from a team and langermankNovember 21, 2023 19:58
@changeset-bot

changeset-botBot commented Nov 21, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2a85556

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

Uh oh! @mperrotti, the image you shared is missing helpful alt text. Check your pull request body.

Alt text is an invisible description that helps screen readers describe images to blind or low-vision users. If you are using markdown to display images, add your alt text inside the brackets of the markdown image.

Learn more about alt text at Basic writing and formatting syntax: images on GitHub Docs.

@github-actions

github-actionsBot commented Nov 21, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.54 KB (+0.15% 🔺)
dist/browser.umd.js105.09 KB (+0.16% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-3963 November 21, 2023 20:03 Inactive
@mperrottimperrotti added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 21, 2023
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 21, 2023
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3963 November 21, 2023 20:53 Inactive
@pksjce

Copy link
Copy Markdown
Contributor

This looks good to me! But I could not find the design decisions for this status type in https://primer.style/ui-patterns/degraded-experiences

@mperrotti

Copy link
Copy Markdown
ContributorAuthor

@pksjce - I don't think it needs to repeated be on both pages.

@github-actions
github-actionsBottemporarily deployed to storybook-preview-3963 November 28, 2023 15:20 Inactive
@mperrotti
mperrotti added this pull request to the merge queueNov 28, 2023
Merged via the queue into main with commit e16927bNov 28, 2023
@mperrotti
mperrotti deleted the mp/unavailable-label branch November 28, 2023 15:32
@primerprimerBot mentioned this pull request Nov 27, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@mperrotti@pksjce@joshblack