Skip to content

Deprecate MarkdownEditor, MarkdownViewer, InlineAutocomplete, and related code - #4027

Merged
iansan5653 merged 12 commits into
mainfrom
deprecate-markdowneditor
Dec 13, 2023
Merged

Deprecate MarkdownEditor, MarkdownViewer, InlineAutocomplete, and related code#4027
iansan5653 merged 12 commits into
mainfrom
deprecate-markdowneditor

Conversation

@iansan5653

@iansan5653iansan5653 commented Dec 7, 2023

Copy link
Copy Markdown
Contributor

Per #3604, we will migrate the following draft/experimental APIs from primer/react to the internal GitHub codebase. Since drafts are not semvered we technically don't have to deprecate them, but I think it's still probably a good idea to follow a deprecation process rather than having them just disappear. So this PR deprecates the components, and then in v37 we can ultimately delete the code. This would be an aggressive timeline for other APIs, but again these are all drafts:

  • MarkdownEditor
  • MardownViewer
  • InlineAutocomplete
  • hooks/useCombobox
  • hooks/useDynamicTextareaHeight
  • hooks/useIgnoreKeyboardActionsWhileComposing
  • hooks/useSafeAsyncCallback
  • hooks/useSyntheticChange
  • hooks/useUnifiedFileSelect, useClickFileSelect, useDropFileSelect, usePasteFileSelect

While some internal APIs will be migrated as well, I've focused on public APIs here.

Side note: I also moved InlineAutocomplete into drafts where it belongs. Not sure how it escaped 🤷.

Changelog

New

Changed

  • Deprecated MarkdownEditor, MardownViewer, InlineAutocomplete, and related hooks

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan

Testing & Reviewing

Merge checklist

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@iansan5653
iansan5653 requested review from a team and pksjceDecember 7, 2023 17:11
@changeset-bot

changeset-botBot commented Dec 7, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 42016cf

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Dec 7, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js105.54 KB (0%)
dist/browser.umd.js106.13 KB (0%)

@siddharthkp
siddharthkp self-requested a review December 7, 2023 17:15
@joshblack

Copy link
Copy Markdown
Member

@iansan5653 just wanted to ask, is usage over in dotcom already migrated over to use the internal version of this component or would it be helpful to add linting to help folks move over to the new package?

@siddharthkp
siddharthkp requested review from broccolinisoup and siddharthkp and removed request for siddharthkpDecember 7, 2023 17:21
@iansan5653

iansan5653 commented Dec 7, 2023

Copy link
Copy Markdown
ContributorAuthor

usage over in dotcom already migrated over to use the internal version of this component or would it be helpful to add linting to help folks move over to the new package?

Not yet. I haven't actually copied the code over to dotcom, but when I do so I will add linting accordingly. I have already added rules to lint specifically against importing @primer/react/drafts/MarkdownEditor since we have CommentBox in dotcom (inflight).

@mperrottimperrotti left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Assuming we don't want to move these to the deprecated package yet, this looks correct to me.

@iansan5653

Copy link
Copy Markdown
ContributorAuthor

Assuming we don't want to move these to the deprecated package yet, this looks correct to me.

Not sure what we would do here, since they are already in the drafts package. I'm open to moving them if you want.

@iansan5653
iansan5653 added this pull request to the merge queueDec 13, 2023
Merged via the queue into main with commit 49f585fDec 13, 2023
@iansan5653
iansan5653 deleted the deprecate-markdowneditor branch December 13, 2023 18:06
This was referenced Dec 13, 2023
siddharthkp added a commit that referenced this pull request Nov 1, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@iansan5653@joshblack@mperrotti