Skip to content

use BaseStyles provider in ConfirmationDialog - #4035

Merged
keithamus merged 2 commits into
mainfrom
use-basestyles-provider-in-confirmationdialog
Dec 8, 2023
Merged

use BaseStyles provider in ConfirmationDialog#4035
keithamus merged 2 commits into
mainfrom
use-basestyles-provider-in-confirmationdialog

Conversation

@keithamus

Copy link
Copy Markdown
Contributor

Closes#4034

Changelog

New

Changed

use in confirm()

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@keithamus
keithamus requested review from a team and mperrottiDecember 8, 2023 16:17
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 100e4da

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js105.16 KB (+0.01% 🔺)
dist/browser.umd.js105.68 KB (+0.01% 🔺)

@keithamus
keithamus added this pull request to the merge queueDec 8, 2023
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Dec 8, 2023
@keithamus
keithamus added this pull request to the merge queueDec 8, 2023
Merged via the queue into main with commit 6497f71Dec 8, 2023
@keithamus
keithamus deleted the use-basestyles-provider-in-confirmationdialog branch December 8, 2023 23:46
@primerprimerBot mentioned this pull request Dec 8, 2023
TylerJDev pushed a commit that referenced this pull request Dec 11, 2023
* use BaseStyles provider in ConfirmationDialog
* changeset
@primerprimerBot mentioned this pull request Dec 19, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

non Portaled confirm() dialogs are unstyled.

2 participants

@keithamus@mperrotti