Skip to content

Add animated prop to ProgressBar - #4061

Merged
mperrotti merged 7 commits into
mainfrom
mp/animated-progress-bar
Dec 20, 2023
Merged

Add animated prop to ProgressBar#4061
mperrotti merged 7 commits into
mainfrom
mp/animated-progress-bar

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Relates to https://github.com/github/primer/issues/2683

Kapture 2023-12-14 at 14 58 16

Changelog

New

  • animated prop on ProgressBar

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@mperrotti
mperrotti requested review from a team and pksjceDecember 14, 2023 20:14
@changeset-bot

changeset-botBot commented Dec 14, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: aa8c0c2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Dec 14, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.87 KB (+0.14% 🔺)
dist/browser.umd.js105.44 KB (+0.14% 🔺)

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great ✨

width: ${props => (props.progress ? `${props.progress}%` : 0)};
background-color: ${get('colors.success.emphasis')};

@media (prefers-reduced-motion: no-preference) {

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Love this 💖

@mperrotti
mperrotti added this pull request to the merge queueDec 20, 2023
Merged via the queue into main with commit f46c071Dec 20, 2023
@mperrotti
mperrotti deleted the mp/animated-progress-bar branch December 20, 2023 21:15
@primerprimerBot mentioned this pull request Dec 20, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@mperrotti@broccolinisoup