Skip to content

chore(deps-dev): bump @types/react from 18.2.21 to 18.2.45 - #4074

Merged
joshblack merged 3 commits into
mainfrom
dependabot/npm_and_yarn/types/react-18.2.45
Dec 19, 2023
Merged

chore(deps-dev): bump @types/react from 18.2.21 to 18.2.45#4074
joshblack merged 3 commits into
mainfrom
dependabot/npm_and_yarn/types/react-18.2.45

Conversation

@dependabot

@dependabotdependabotBot commented on behalf of githubDec 18, 2023

Copy link
Copy Markdown
Contributor

Bumps @types/react from 18.2.21 to 18.2.45.

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

@dependabot
dependabotBot requested review from a team and joshblackDecember 18, 2023 02:26
@dependabotdependabotBot added dependencies Pull requests that update a dependency file skip changeset This change does not need a changelog labels Dec 18, 2023
@changeset-bot

changeset-botBot commented Dec 18, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dc8690f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dependabot
dependabotBotforce-pushed the dependabot/npm_and_yarn/types/react-18.2.45 branch from 842e7a6 to 6b839daCompareDecember 19, 2023 03:12
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4074 December 19, 2023 03:16 Inactive
@dependabot
dependabotBotforce-pushed the dependabot/npm_and_yarn/types/react-18.2.45 branch 2 times, most recently from ab79c12 to e2f8525CompareDecember 19, 2023 18:42
Bumps [@types/react](https://github.com/DefinitelyTyped/DefinitelyTyped/tree/HEAD/types/react) from 18.2.21 to 18.2.45.
- [Release notes](https://github.com/DefinitelyTyped/DefinitelyTyped/releases)
- [Commits](https://github.com/DefinitelyTyped/DefinitelyTyped/commits/HEAD/types/react)
---
updated-dependencies:
- dependency-name: "@types/react"
dependency-type: direct:development
update-type: version-update:semver-patch
...
Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabotBotforce-pushed the dependabot/npm_and_yarn/types/react-18.2.45 branch from e2f8525 to 4c7f477CompareDecember 19, 2023 18:43
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4074 December 19, 2023 18:48 Inactive
@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.74 KB (0%)
dist/browser.umd.js105.29 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4074 December 19, 2023 19:22 Inactive
@joshblack
joshblack added this pull request to the merge queueDec 19, 2023
Merged via the queue into main with commit 06c7c6cDec 19, 2023
@joshblack
joshblack deleted the dependabot/npm_and_yarn/types/react-18.2.45 branch December 19, 2023 19:34
@primerprimerBot mentioned this pull request Dec 19, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependenciesPull requests that update a dependency fileskip changesetThis change does not need a changelog

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@joshblack