Skip to content

Tooltip2: Update direction mapping - #4098

Merged
broccolinisoup merged 2 commits into
mainfrom
bs/tooltip-direction
Jan 12, 2024
Merged

Tooltip2: Update direction mapping#4098
broccolinisoup merged 2 commits into
mainfrom
bs/tooltip-direction

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Dec 21, 2023

Copy link
Copy Markdown
Member

Just realised that the Tooltip v2 se, sw, ne, nw directions are not correct.

What we have vs What is expected.

Changelog

New

Changed

Updated the align property values in the mapping objects based on the info from the behaviours:

 /**
* Determines how the floating element should align with the anchor element. If
* set to "start", the floating element's first edge (top or left) will align
* with the anchor element's first edge. If set to "center", the floating
* element will be centered along the axis of the anchor edge. If set to "end",
* the floating element's last edge will align with the anchor element's last edge.
*/

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

All Directions story is a good one to test

Merge checklist

@broccolinisoup
broccolinisoup requested review from a team and pksjceDecember 21, 2023 01:36
@changeset-bot

changeset-botBot commented Dec 21, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c66bc61

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@broccolinisoupbroccolinisoup changed the title yTooltip2: Update direction mappingDec 21, 2023
@github-actions

github-actionsBot commented Dec 21, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.95 KB (+0.01% 🔺)
dist/browser.umd.js105.52 KB (0%)

@mperrottimperrotti left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

It would also be nice to follow this up to a change in the Playground story that lets us pick a direction from a predefined list of options instead of accepting any string.

@broccolinisoup

broccolinisoup commented Dec 22, 2023

Copy link
Copy Markdown
MemberAuthor

It would also be nice to follow this up to a change in the Playground story that lets us pick a direction from a predefined list of options instead of accepting any string.

Oh great feedback! I didn't even realise that it was set as string, sorry my bad. I updated it 🤗

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4098 December 22, 2023 04:40 Inactive
@broccolinisoup
broccolinisoup added this pull request to the merge queueJan 12, 2024
Merged via the queue into main with commit dc97a9bJan 12, 2024
@broccolinisoup
broccolinisoup deleted the bs/tooltip-direction branch January 12, 2024 05:36
@primerprimerBot mentioned this pull request Jan 11, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@broccolinisoup@mperrotti