Skip to content

Update Link inline prop description and deprecate underline - #4135

Merged
dipree merged 7 commits into
mainfrom
dipree/update-link-inline-prop-description
Jan 12, 2024
Merged

Update Link inline prop description and deprecate underline#4135
dipree merged 7 commits into
mainfrom
dipree/update-link-inline-prop-description

Conversation

@dipree

Copy link
Copy Markdown
Contributor

https://github.com/github/primer/issues/2099

Updates the Link inline property description and set the underline property to "deprecated".

@changeset-bot

changeset-botBot commented Jan 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5deac0e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jan 10, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.96 KB (0%)
dist/browser.umd.js105.53 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4135 January 10, 2024 09:06 Inactive
@dipree
dipree requested a review from joshblackJanuary 10, 2024 09:37
@dipree
dipree marked this pull request as ready for review January 11, 2024 08:08
@dipree
dipree requested a review from a teamJanuary 11, 2024 08:08

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for doing this! Just left a comment on where @deprecated could live, hope it makes sense and let me know if you have any questions!

Comment threadsrc/Link/Link.tsx Outdated
dipreeand others added 2 commits January 11, 2024 17:16
Co-authored-by: Josh Black <joshblack@github.com>
@dipree

Copy link
Copy Markdown
ContributorAuthor

@joshblack I see how that works now, thanks and updated.

@dipree
dipree added this pull request to the merge queueJan 12, 2024
Merged via the queue into main with commit c2b069cJan 12, 2024
@dipree
dipree deleted the dipree/update-link-inline-prop-description branch January 12, 2024 18:45
@primerprimerBot mentioned this pull request Jan 12, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@dipree@joshblack