Uh oh!
There was an error while loading. Please reload this page.
refactor(project): drop focus-visible polyfill - #4168
Conversation
🦋 Changeset detectedLatest commit: 9cb86bc The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
size-limit report 📦
|
TylerJDev
commented
Jan 22, 2024
I think this might affect the changes that could happen in #4166. Does this mean we'd need to find a new solution to fix the referenced bug? 🤔 |
There was a problem hiding this comment.
🎉
Let's test this with dotcom to be extra extra sure! For example, do these 3 instances depend on primer brining the polyfill?
I selected a minor release here but could also see this considered a breaking change since we don't explicitly call out our browser compatability (unless we indicate that it should match GitHub's?)
Happy with minor
joshblack
commented
Feb 8, 2024
Closing as we currently need this polyfill for some odd Safari behavior 🤔 |
In our maintainer sync today, it came up that we no longer need to polyfill
focus-visiblebased on GitHub's browser support range. This PR removes this polyfill and any references to the explicit.focus-visibleclassChangelog
New
Changed
.focus-visibleto just:focus-visibleRemoved
focus-visiblepolyfillRollout strategy
I selected a minor release here but could also see this considered a breaking change since we don't explicitly call out our browser compatability (unless we indicate that it should match GitHub's?)
Testing & Reviewing
:focus-visible. We should make sure no remaining.focus-visibleselectors remain or are in use