Skip to content

Checkbox: Add cursor:not-allowed to unchecked disabled state - #4176

Merged
siddharthkp merged 4 commits into
mainfrom
disabled-checkbox-cursor
Jan 30, 2024
Merged

Checkbox: Add cursor:not-allowed to unchecked disabled state#4176
siddharthkp merged 4 commits into
mainfrom
disabled-checkbox-cursor

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Jan 24, 2024

Copy link
Copy Markdown
Member

@changeset-bot

changeset-botBot commented Jan 24, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: fa29ec8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp changed the title Checkbox: Add cursor:not-allowed to unchecked stateCheckbox: Add cursor:not-allowed to unchecked disabled stateJan 24, 2024
@github-actions

github-actionsBot commented Jan 24, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js113.23 KB (+0.01% 🔺)
dist/browser.umd.js113.88 KB (+0.01% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4176 January 24, 2024 12:35 Inactive
@siddharthkpsiddharthkp self-assigned this Jan 24, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review January 24, 2024 12:46
@siddharthkp
siddharthkp requested review from a team, broccolinisoup and joshblack and removed request for broccolinisoupJanuary 24, 2024 12:46
@siddharthkp
siddharthkpforce-pushed the disabled-checkbox-cursor branch from 77bece7 to fa29ec8CompareJanuary 30, 2024 17:55
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4176 January 30, 2024 17:59 Inactive
@siddharthkp
siddharthkp added this pull request to the merge queueJan 30, 2024
Merged via the queue into main with commit b8b090bJan 30, 2024
@siddharthkp
siddharthkp deleted the disabled-checkbox-cursor branch January 30, 2024 18:13
@primerprimerBot mentioned this pull request Jan 30, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Checkbox label gets not-allowed cursor but input itself still has pointer

2 participants

@siddharthkp@joshblack