Skip to content

Add missing CSS vars to legacy-theme - #4193

Merged
langermank merged 7 commits into
mainfrom
css-var-take-2
Jan 29, 2024
Merged

Add missing CSS vars to legacy-theme#4193
langermank merged 7 commits into
mainfrom
css-var-take-2

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Adding in some values I missed in the previous PR.

@langermank
langermank requested review from a team and pksjceJanuary 29, 2024 20:20
@changeset-bot

changeset-botBot commented Jan 29, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d10a6f6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jan 29, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js113.23 KB (-0.53% 🔽)
dist/browser.umd.js113.88 KB (-0.52% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4193 January 29, 2024 20:25 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 29, 2024

@mperrottimperrotti left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

The changes look correct to me. Since I'm not 100% certain about all of our available tokens, I'm can't say with confidence that you didn't miss anything.

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 29, 2024
@mperrotti

Copy link
Copy Markdown
Contributor

I just realized the visual regression tests' snapshots look a little off. Primary buttons are no longer green, and the button padding seems to have disappeared.

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 29, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4193 January 29, 2024 21:36 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 29, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4193 January 29, 2024 21:43 Inactive
@langermank
langermank added this pull request to the merge queueJan 29, 2024
Merged via the queue into main with commit c4193d6Jan 29, 2024
@langermank
langermank deleted the css-var-take-2 branch January 29, 2024 21:56
@primerprimerBot mentioned this pull request Jan 29, 2024
pksjce pushed a commit that referenced this pull request Feb 8, 2024
* oof
* Create thirty-mirrors-switch.md
* test(vrt): update snapshots
* fix missing var
* fix parens
* test(vrt): update snapshots
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
github-merge-queueBot pushed a commit that referenced this pull request Feb 12, 2024
* Tentative focus fix in action list item
* Update snapshots
* Create purple-bees-warn.md
* chore(deps-dev): bump @babel/cli from 7.22.10 to 7.23.9 (#4189)
Bumps [@babel/cli](https://github.com/babel/babel/tree/HEAD/packages/babel-cli) from 7.22.10 to 7.23.9.
- [Release notes](https://github.com/babel/babel/releases)
- [Changelog](https://github.com/babel/babel/blob/main/CHANGELOG.md)
- [Commits](https://github.com/babel/babel/commits/v7.23.9/packages/babel-cli)
---
updated-dependencies:
- dependency-name: "@babel/cli"
dependency-type: direct:development
update-type: version-update:semver-minor
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* chore(deps-dev): bump terser from 5.17.6 to 5.27.0 (#4187)
Bumps [terser](https://github.com/terser/terser) from 5.17.6 to 5.27.0.
- [Changelog](https://github.com/terser/terser/blob/master/CHANGELOG.md)
- [Commits](terser/terser@v5.17.6...v5.27.0)
---
updated-dependencies:
- dependency-name: terser
dependency-type: direct:development
update-type: version-update:semver-minor
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* chore(deps-dev): bump @storybook/source-loader from 7.6.7 to 7.6.10 (#4186)
Bumps [@storybook/source-loader](https://github.com/storybookjs/storybook/tree/HEAD/code/lib/source-loader) from 7.6.7 to 7.6.10.
- [Release notes](https://github.com/storybookjs/storybook/releases)
- [Changelog](https://github.com/storybookjs/storybook/blob/next/CHANGELOG.md)
- [Commits](https://github.com/storybookjs/storybook/commits/v7.6.10/code/lib/source-loader)
---
updated-dependencies:
- dependency-name: "@storybook/source-loader"
dependency-type: direct:development
update-type: version-update:semver-patch
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* Improve aria-labels for example icon buttons and add guidance (#4191)
* Improve aria-labels for example icon buttons and add guidance
* test(vrt): update snapshots
---------
Co-authored-by: joshblack <joshblack@users.noreply.github.com>
* chore(deps): bump @github/combobox-nav from 2.1.7 to 2.3.1 (#4185)
* chore(deps): bump @github/combobox-nav from 2.1.7 to 2.3.1
Bumps [@github/combobox-nav](https://github.com/github/combobox-nav) from 2.1.7 to 2.3.1.
- [Release notes](https://github.com/github/combobox-nav/releases)
- [Commits](github/combobox-nav@v2.1.7...v2.3.1)
---
updated-dependencies:
- dependency-name: "@github/combobox-nav"
dependency-type: direct:production
update-type: version-update:semver-minor
...
Signed-off-by: dependabot[bot] <support@github.com>
* test: add fallback for Element#scrollIntoView
* chore: fix eslint warning
---------
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* Add missing CSS vars to legacy-theme (#4193)
* oof
* Create thirty-mirrors-switch.md
* test(vrt): update snapshots
* fix missing var
* fix parens
* test(vrt): update snapshots
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
* Version Packages (#4183)
Co-authored-by: github-actions[bot] <github-actions[bot]@users.noreply.github.com>
---------
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
Co-authored-by: Owen Niblock <owenniblock@github.com>
Co-authored-by: joshblack <joshblack@users.noreply.github.com>
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: primer[bot] <119360173+primer[bot]@users.noreply.github.com>
Co-authored-by: github-actions[bot] <github-actions[bot]@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@mperrotti