Skip to content

SelectPanel2: Export props + pass props - #4198

Merged
siddharthkp merged 6 commits into
mainfrom
draft-selectpanel-export-props
Feb 2, 2024
Merged

SelectPanel2: Export props + pass props#4198
siddharthkp merged 6 commits into
mainfrom
draft-selectpanel-export-props

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Jan 30, 2024

Copy link
Copy Markdown
Member

Rollout strategy

  • Patch release

Testing & Reviewing

Merge checklist

@siddharthkpsiddharthkp added the skip changeset This change does not need a changelog label Jan 30, 2024
@changeset-bot

changeset-botBot commented Jan 30, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 344b48b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp removed the skip changeset This change does not need a changelog label Jan 30, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review January 30, 2024 19:11
@siddharthkp
siddharthkp requested a review from a teamJanuary 30, 2024 19:11
@github-actions

github-actionsBot commented Jan 30, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js113.24 KB (0%)
dist/browser.umd.js113.88 KB (0%)

@ayy-bc

Copy link
Copy Markdown
Contributor

@siddharthkp we would also need the SelectPanelSecondaryActionProps to be exported to use it with the item picker for cases where we need secondary actions (related: https://github.com/github/collaboration-workflows-flex/issues/423, https://github.com/github/github/pull/310856/files).

Do you want me to create another issue to track that or can that also go as a part of this PR?

'100%': {transform: 'scale(1, 1)'},
},
}}
{...props}

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@siddharthkp Does the typescript on props also allow StyledOverlay props to be passed through?

For context, I'm asking case where the prop maxHeight is needed to fix a styling issue with the overlay height: https://github.com/github/collaboration-workflows-flex/issues/449

<SelectPanel
height="fit-content"
maxHeight="large"
...
/>

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Does the typescript on props also allow StyledOverlay props to be passed through?

Not yet!

I'm asking case where the prop maxHeight is needed to fix a styling issue with the overlay height: https://github.com/github/collaboration-workflows-flex/issues/449

That's an interesting issue, not sure if maxHeight is the solution. Going to reply on the issue first itself

@siddharthkp

Copy link
Copy Markdown
MemberAuthor

@ayy-bc Sure thing, I can add them to this PR

Merged via the queue into main with commit 65bde27Feb 2, 2024
@siddharthkp
siddharthkp deleted the draft-selectpanel-export-props branch February 2, 2024 19:33
@primerprimerBot mentioned this pull request Feb 2, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants

@siddharthkp@ayy-bc@broccolinisoup@joshblack@JelloBagel