Skip to content

Change Tooltip border-radius from 3px to 6px - #4204

Merged
dylanatsmith merged 6 commits into
mainfrom
dylanatsmith-patch-1
Feb 6, 2024
Merged

Change Tooltip border-radius from 3px to 6px#4204
dylanatsmith merged 6 commits into
mainfrom
dylanatsmith-patch-1

Conversation

@dylanatsmith

Copy link
Copy Markdown
Member

The Tooltip border-radius is inconsistent between Rails and React components. This is obvious where the two live side-by-side, like the global Copilot header button.

This PR assumes Rails is correct and changes React to match.

Screen.Recording.2024-02-01.at.11.59.58.mov
Screen.Recording.2024-02-01.at.12.00.24.mov

Changelog

New

Changed

Tooltip border-radius

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@dylanatsmithdylanatsmith self-assigned this Feb 1, 2024
@dylanatsmith
dylanatsmith requested review from a team and langermankFebruary 1, 2024 12:00
@changeset-bot

changeset-botBot commented Feb 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a4edb42

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 1, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js113.22 KB (0%)
dist/browser.umd.js113.86 KB (0%)

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM! Will defer to @langermank for a visual review 👀

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Heads up we're about to refactor this design (CC @broccolinisoup) but probably no harm in merging this?

@broccolinisoup

Copy link
Copy Markdown
Member

Heads up we're about to refactor this design (CC @broccolinisoup) but probably no harm in merging this?

Thanks for the heads up @langermank !! No problem at all, I'll just take that line back in this PR

@dylanatsmith

Copy link
Copy Markdown
MemberAuthor

@langermank@broccolinisoup Can either of you help me with the failing test on this? I’m not sure where/how to update it.

@langermank

Copy link
Copy Markdown
Contributor

@dylanatsmith try running npm test -- -u and push the snapshot changes up

Merged via the queue into main with commit da4469dFeb 6, 2024
@dylanatsmith
dylanatsmith deleted the dylanatsmith-patch-1 branch February 6, 2024 20:31
@primerprimerBot mentioned this pull request Feb 6, 2024
@dylanatsmith

Copy link
Copy Markdown
MemberAuthor

Thanks @joshblack@langermank 🙏

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@dylanatsmith@broccolinisoup@langermank@joshblack