Skip to content

fix(theme): add font stack as fallback to custom property - #4208

Merged
joshblack merged 3 commits into
mainfrom
fix/update-font-family-mono
Feb 1, 2024
Merged

fix(theme): add font stack as fallback to custom property#4208
joshblack merged 3 commits into
mainfrom
fix/update-font-family-mono

Conversation

@joshblack

Copy link
Copy Markdown
Member

Update the value for mono in our theme. The font stack is now the fallback value of the CSS Custom Property

Changelog

New

Changed

  • Update mono font family value in theme

Removed

Rollout strategy

  • Patch release

Testing & Reviewing

  • Verify code snippets and other instances that use a mono value from the theme now render with the correct font family

@joshblack
joshblack requested review from a team and siddharthkpFebruary 1, 2024 19:07
@changeset-bot

changeset-botBot commented Feb 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a95cec4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblackjoshblack changed the title Fix/update font family monofix(theme): add font stack as fallback to custom propertyFeb 1, 2024
@github-actions

github-actionsBot commented Feb 1, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js113.24 KB (+0.01% 🔺)
dist/browser.umd.js113.88 KB (+0.01% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4208 February 1, 2024 19:19 Inactive
@joshblack
joshblack added this pull request to the merge queueFeb 1, 2024
Merged via the queue into main with commit 2217cefFeb 1, 2024
@joshblack
joshblack deleted the fix/update-font-family-mono branch February 1, 2024 19:32
@primerprimerBot mentioned this pull request Jan 30, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@langermank