Skip to content

SelectPanel2: Automatically adjust height based on contents - #4245

Merged
siddharthkp merged 6 commits into
mainfrom
drafts-selectpanel-autoheight
Feb 12, 2024
Merged

SelectPanel2: Automatically adjust height based on contents#4245
siddharthkp merged 6 commits into
mainfrom
drafts-selectpanel-autoheight

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Feb 9, 2024

Copy link
Copy Markdown
Member

I was under the false impression that SelectPanel should retain it's height, but we do want it to resize with contents!

Before:

selectpanel-static.mov

After:

selectpanel-auto-resize.mov

@siddharthkpsiddharthkp self-assigned this Feb 9, 2024
@siddharthkp
siddharthkp requested a review from a teamFebruary 9, 2024 16:06
@changeset-bot

changeset-botBot commented Feb 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e9bff98

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp changed the title Experimental SelectPanel2: Automatically adjust height based on contentsSelectPanel2: Automatically adjust height based on contentsFeb 9, 2024
@github-actions

github-actionsBot commented Feb 9, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.38 KB (0%)
packages/react/dist/browser.umd.js114.03 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4245 February 12, 2024 15:48 Inactive
@siddharthkp
siddharthkp added this pull request to the merge queueFeb 12, 2024
Merged via the queue into main with commit d79a419Feb 12, 2024
@siddharthkp
siddharthkp deleted the drafts-selectpanel-autoheight branch February 12, 2024 16:14
@primerprimerBot mentioned this pull request Feb 12, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* replace height with maxHeight
* adjust height of loading & message based on container
* remove redundant height from stories
* Create strange-starfishes-work.md
* fix bad merge
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@siddharthkp@broccolinisoup@joshblack