Skip to content

Add a new entry point as next and update draft Tooltip to export it from next - #4250

Merged
broccolinisoup merged 10 commits into
mainfrom
next-entry-point-tooltip
Feb 18, 2024
Merged

Add a new entry point as next and update draft Tooltip to export it from next#4250
broccolinisoup merged 10 commits into
mainfrom
next-entry-point-tooltip

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Feb 10, 2024

Copy link
Copy Markdown
Member

Based on the Tooltip v2 adoption discussion, this PR adds a new entry point as next and export Tooltip from it with an updated documentation. The discussion is still open for feedback for docs consolidation, so please feel free to share any ideas/thoughts/concerns 🙌🏻

Changelog

New

  • next entry point

Changed

  • Renamed the draft/experimental Tooltip in docs and storybook to become Tooltip v2

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

  • Is it clear why there are two version of Tooltips under the components?
  • Is it visible enough that the import path of Tooltip v2 is different than the usual import path?
  • Checking out the Tooltip and Tooltip v2 documentation and storybook and make sure they reflect the right version of the component
  • If possible testing the export would be great. i.e import {Tooltip} from '@primer/react/next' and make sure it imports the component properly.
  • Integration PR (Github stuff only)

Merge checklist

@changeset-bot

changeset-botBot commented Feb 10, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6a0faea

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@broccolinisoupbroccolinisoup changed the title Add a new entry point as and update draft Tooltip to export it from nextAdd a new entry point as next and update draft Tooltip to export it from nextFeb 10, 2024
@github-actions

github-actionsBot commented Feb 10, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.37 KB (-0.01% 🔽)
packages/react/dist/browser.umd.js114.03 KB (+0.01% 🔺)

Comment threaddocs/content/TooltipV2.mdx Outdated

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🙌 This is coming together!

All the feedback is optional except for the docs preview. "Requested changes", feel free to DM me if I don't look at the notification once you make the changes

Comment threaddocs/content/Tooltip.mdx
Comment threaddocs/content/TooltipV2.mdx Outdated
Comment threaddocs/src/@primer/gatsby-theme-doctocat/nav.yml

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM for the next entrypoint! 👍 Will defer to @siddharthkp's comments for stories/location 👀

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4250 February 14, 2024 05:26 Inactive
@broccolinisoupbroccolinisoup added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 14, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 14, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4250 February 14, 2024 06:10 Inactive

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Docs work now! Ship it :shipit:

Comment threaddocs/content/TooltipV2.mdx
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4250 February 16, 2024 05:06 Inactive
@broccolinisoup
broccolinisoup added this pull request to the merge queueFeb 18, 2024
Merged via the queue into main with commit f064534Feb 18, 2024
@broccolinisoup
broccolinisoup deleted the next-entry-point-tooltip branch February 18, 2024 23:05
@primerprimerBot mentioned this pull request Feb 18, 2024
broccolinisoup added a commit that referenced this pull request Feb 20, 2024
… from next (#4250)
* Add a new entry point as and update draft Tooltip to export it from next
* fix linting n add changeset
* move the src code out of drafts and update docs
* fix folder path and storyname
* test(vrt): update snapshots
* update tooltip's position to be absolute
* Update docs/content/TooltipV2.mdx
---------
Co-authored-by: broccolinisoup <broccolinisoup@users.noreply.github.com>
@primerprimerBot mentioned this pull request Feb 28, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
… from next (#4250)
* Add a new entry point as and update draft Tooltip to export it from next
* fix linting n add changeset
* move the src code out of drafts and update docs
* fix folder path and storyname
* test(vrt): update snapshots
* update tooltip's position to be absolute
* Update docs/content/TooltipV2.mdx
---------
Co-authored-by: broccolinisoup <broccolinisoup@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@broccolinisoup@siddharthkp@joshblack