Skip to content

SelectPanel2: Submit panel when an item is selected with Enter - #4265

Merged
siddharthkp merged 2 commits into
mainfrom
drafts-selectpanel-enter-to-submit
Feb 13, 2024
Merged

SelectPanel2: Submit panel when an item is selected with Enter#4265
siddharthkp merged 2 commits into
mainfrom
drafts-selectpanel-enter-to-submit

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Feb 13, 2024

Copy link
Copy Markdown
Member
  • Pressing Space selects and deselects the focused item
  • Pressing Enter selects and deselects the focused item AND submits and closes panel
selectpanel-enter.mp4

Known bug: Pressing Enter in the search input also submits the panel which it should not, will update that after #4199 is merged

@siddharthkpsiddharthkp added the patch release bug fixes, docs, housekeeping label Feb 13, 2024
@siddharthkpsiddharthkp self-assigned this Feb 13, 2024
@changeset-bot

changeset-botBot commented Feb 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 647bdb8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkp
siddharthkp marked this pull request as ready for review February 13, 2024 16:12
@siddharthkp
siddharthkp requested a review from a teamFebruary 13, 2024 16:12
@github-actions

github-actionsBot commented Feb 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.37 KB (-0.01% 🔽)
packages/react/dist/browser.umd.js114.01 KB (-0.01% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4265 February 13, 2024 16:15 Inactive
@siddharthkp
siddharthkp added this pull request to the merge queueFeb 13, 2024
Merged via the queue into main with commit 19c716fFeb 13, 2024
@siddharthkp
siddharthkp deleted the drafts-selectpanel-enter-to-submit branch February 13, 2024 16:37
@primerprimerBot mentioned this pull request Feb 13, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* submit on Enter
* Create polite-rockets-yawn.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

patch releasebug fixes, docs, housekeeping

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@joshblack