Skip to content

SelectPanel2: Responsive variants - #4277

Merged
siddharthkp merged 12 commits into
mainfrom
drafts-selectpanel-fullscreen
Mar 19, 2024
Merged

SelectPanel2: Responsive variants#4277
siddharthkp merged 12 commits into
mainfrom
drafts-selectpanel-fullscreen

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Feb 15, 2024

Copy link
Copy Markdown
Member
- variant?: 'anchored' | 'modal'+ variant?: 'anchored' | 'modal' | ResponsiveValue<'anchored' | 'modal', 'full-screen' | 'bottom-sheet'>

which let's you do things like

<SelectPanelvariant={{narrow: 'bottom-sheet',regular: 'modal'}}>

Play around with it on storybook or watch this video:

selectpanel-responsive.mov

@changeset-bot

changeset-botBot commented Feb 15, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4fba95a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4277 February 15, 2024 16:05 Inactive
@siddharthkpsiddharthkp added bug Something isn't working 💓collab a vibrant hub of collaboration and removed bug Something isn't working 💓collab a vibrant hub of collaboration labels Feb 27, 2024
@github-actions

github-actionsBot commented Mar 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js87.51 KB (0%)
packages/react/dist/browser.umd.js87.74 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4277 March 12, 2024 16:22 Inactive
@siddharthkpsiddharthkp self-assigned this Mar 12, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review March 12, 2024 16:27
@siddharthkp
siddharthkp requested a review from a team as a code ownerMarch 12, 2024 16:27
@siddharthkp
siddharthkp added this pull request to the merge queueMar 18, 2024
@joshblack
joshblack removed this pull request from the merge queue due to a manual request Mar 18, 2024
@siddharthkp
siddharthkp added this pull request to the merge queueMar 19, 2024
Merged via the queue into main with commit 69915d9Mar 19, 2024
@siddharthkp
siddharthkp deleted the drafts-selectpanel-fullscreen branch March 19, 2024 09:53
@primerprimerBot mentioned this pull request Mar 19, 2024
@siddharthkpsiddharthkp changed the title SelectPanel2: Responsive variantsexperimental/SelectPanel2: Responsive variantsMar 19, 2024
@siddharthkpsiddharthkp changed the title experimental/SelectPanel2: Responsive variantsexperimental/SelectPanel: Responsive variantsMar 19, 2024
@siddharthkpsiddharthkp changed the title experimental/SelectPanel: Responsive variantsSelectPanel2: Responsive variantsMar 19, 2024
@kitze

kitze commented Apr 3, 2024

Copy link
Copy Markdown

@siddharthkp di.. did you just resize your browser aaaa

lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* cleanup: outdated comment
* full screen based on media query
* automated responsive button size
* use ResponsiveVariants
* add responsive story
* add bottom-sheet styles
* Create proud-ears-travel.md
* add matchMedia mock to test-helpers
* Revert "add matchMedia mock to test-helpers"
This reverts commit d71e0f6.
* mock matchMedia for SelectPanel test
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@siddharthkp@kitze@joshblack