Skip to content

SelectPanel2: Cancel + close panel when user clicks outside - #4294

Merged
siddharthkp merged 6 commits into
mainfrom
drafts-selectpanel-click-outside
Feb 22, 2024
Merged

SelectPanel2: Cancel + close panel when user clicks outside#4294
siddharthkp merged 6 commits into
mainfrom
drafts-selectpanel-click-outside

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Feb 21, 2024

Copy link
Copy Markdown
Member

  1. selectionVariant=instant
  • Clicking on option selects it, calls onSubmit and closes the menu
  • Clicking cancel button or close button calls onCancel and closes the menu
  • Clicking outside calls onCancel and closes the menu
  1. selectionVariant=multiple | single
  • Clicking on option selects it.
  • Clicking save calls onSubmit and closes the menu
  • Clicking cancel button or close button calls onCancel and closes the menu
  • Clicking outside calls onCancel and closes the menu

@changeset-bot

changeset-botBot commented Feb 21, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 517a84f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp self-assigned this Feb 21, 2024
@github-actions

github-actionsBot commented Feb 21, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.47 KB (0%)
packages/react/dist/browser.umd.js114.14 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4294 February 22, 2024 16:00 Inactive
@siddharthkp
siddharthkp added this pull request to the merge queueFeb 22, 2024
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Feb 22, 2024
@siddharthkp
siddharthkp added this pull request to the merge queueFeb 22, 2024
Merged via the queue into main with commit 5ca5c0aFeb 22, 2024
@siddharthkp
siddharthkp deleted the drafts-selectpanel-click-outside branch February 22, 2024 16:40
@primerprimerBot mentioned this pull request Feb 20, 2024
@siddharthkpsiddharthkp changed the title experimental/SelectPanel: Cancel + close panel when user clicks outsideSelectPanel2: Cancel + close panel when user clicks outsideMar 19, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
…de (#4294)
* close on clicking outside
* add link to decision log
* Create nervous-dogs-change.md
* add onCancel to all examples
* oopsie!
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@joshblack