Skip to content

SelectPanel2: Move focus to first item if there is no filter input - #4320

Merged
siddharthkp merged 3 commits into
mainfrom
draft-selectpanel-minimal-initial-focus
Mar 7, 2024
Merged

SelectPanel2: Move focus to first item if there is no filter input#4320
siddharthkp merged 3 commits into
mainfrom
draft-selectpanel-minimal-initial-focus

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Feb 28, 2024

Copy link
Copy Markdown
Member
BeforeAfter
initial focus is on the first interactive element, the close buttoninitial focus is on the first list element
initial focus is on the first interactive element, the close buttoninitial focus is on the first list element

@changeset-bot

changeset-botBot commented Feb 28, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 66b5d3a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp self-assigned this Feb 28, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review February 28, 2024 15:16
@siddharthkp
siddharthkp requested a review from a teamFebruary 28, 2024 15:16
@github-actions

github-actionsBot commented Feb 28, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.71 KB (0%)
packages/react/dist/browser.umd.js114.38 KB (0%)

@siddharthkp
siddharthkp requested a review from a team as a code ownerMarch 7, 2024 11:23
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4320 March 7, 2024 11:27 Inactive
@siddharthkp
siddharthkp added this pull request to the merge queueMar 7, 2024
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Mar 7, 2024
@siddharthkp
siddharthkp added this pull request to the merge queueMar 7, 2024
Merged via the queue into main with commit 7cb3464Mar 7, 2024
@siddharthkp
siddharthkp deleted the draft-selectpanel-minimal-initial-focus branch March 7, 2024 11:54
@primerprimerBot mentioned this pull request Mar 7, 2024
@siddharthkpsiddharthkp changed the title experimental/SelectPanel: Move focus to first item if there is no filter inputSelectPanel2: Move focus to first item if there is no filter inputMar 19, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
…ter input (#4320)
* move focus to list if there is no input
* Create neat-toes-build.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@joshblack