Skip to content

Add word-break to ActionList - #4331

Merged
langermank merged 10 commits into
mainfrom
action-list-break-word
Apr 15, 2024
Merged

Add word-break to ActionList#4331
langermank merged 10 commits into
mainfrom
action-list-break-word

Conversation

@langermank

@langermanklangermank commented Feb 29, 2024

Copy link
Copy Markdown
Contributor

Allow long strings without natural breaks to break-word in ActionList item.

Before:
dont break word

After:
break word

@changeset-bot

changeset-botBot commented Feb 29, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1c02482

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 29, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js88 KB (+0.05% 🔺)
packages/react/dist/browser.umd.js88.25 KB (-0.08% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4331 February 29, 2024 16:54 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 29, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 29, 2024
@siddharthkp

siddharthkp commented Mar 1, 2024

Copy link
Copy Markdown
Member

Surprising that this doesn't update any visual tests! Do you mind adding one just so that we have it for the future!

Side note: This is probably confusing, but the update-snapshots label only updates visual testing screenshots not jest snapshots. For that you need to run npm run test:update

@langermank

Copy link
Copy Markdown
ContributorAuthor

Thanks @siddharthkp, good call. I added a story

break word

@langermank
langermank marked this pull request as ready for review April 11, 2024 22:58
@langermank
langermank requested a review from a team as a code ownerApril 11, 2024 22:58
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4331 April 11, 2024 23:02 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 12, 2024
@langermank
langermank requested a review from a team as a code ownerApril 12, 2024 22:02
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 12, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4331 April 12, 2024 22:05 Inactive
@langermank
langermank enabled auto-merge April 15, 2024 18:12
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4331 April 15, 2024 18:17 Inactive
@langermank
langermank added this pull request to the merge queueApr 15, 2024
Merged via the queue into main with commit 77846c4Apr 15, 2024
@langermank
langermank deleted the action-list-break-word branch April 15, 2024 18:31
@primerprimerBot mentioned this pull request Apr 15, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@langermank@siddharthkp@mperrotti