Skip to content

markdown toolbar - #4345

Closed
keithamus wants to merge 13 commits into
mainfrom
markdown-toolbar-draft
Closed

markdown toolbar#4345
keithamus wants to merge 13 commits into
mainfrom
markdown-toolbar-draft

Conversation

@keithamus

@keithamus keithamus commented Mar 4, 2024

Copy link
Copy Markdown
Contributor

Closes https://github.com/github/primer/issues/3107

Adds a new MarkdownToolbar component leveraging <markdown-toolbar> to allow for a <textarea> to take formatting buttons.

Changelog

  • Adds a new MarkdownToolbar draft component.

New

  • MarkdownToolbar Compoent

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-bot Bot commented Mar 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f3da8f2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@primer/react Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@keithamus
keithamus force-pushed the markdown-toolbar-draft branch from 2bf75ab to c5b9907 Compare March 6, 2024 11:57
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 6, 2024 12:01 Inactive
@keithamus
keithamus force-pushed the markdown-toolbar-draft branch from c5b9907 to c17f6b7 Compare March 7, 2024 17:22
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 7, 2024 17:26 Inactive
@keithamus
keithamus force-pushed the markdown-toolbar-draft branch from c17f6b7 to e15b97e Compare March 8, 2024 16:32
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 8, 2024 16:36 Inactive
@keithamus
keithamus force-pushed the markdown-toolbar-draft branch from e15b97e to eaaf548 Compare March 8, 2024 16:38
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 8, 2024 16:41 Inactive
@keithamus
keithamus force-pushed the markdown-toolbar-draft branch from eaaf548 to ad1d3c9 Compare March 8, 2024 16:46
@github-actions

github-actions Bot commented Mar 8, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

Path Size
packages/react/dist/browser.esm.js 87.54 KB (0%)
packages/react/dist/browser.umd.js 87.81 KB (0%)

@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 8, 2024 16:50 Inactive
@keithamus
keithamus force-pushed the markdown-toolbar-draft branch from ad1d3c9 to c9a8cad Compare March 8, 2024 17:14
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 8, 2024 17:17 Inactive
@keithamus
keithamus force-pushed the markdown-toolbar-draft branch from c9a8cad to 4328979 Compare March 8, 2024 17:25
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 8, 2024 17:29 Inactive
@keithamus
keithamus force-pushed the markdown-toolbar-draft branch from 4328979 to 436bc7f Compare March 8, 2024 17:59
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 8, 2024 18:02 Inactive
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 12, 2024 12:58 Inactive
@keithamus keithamus added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 13, 2024
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 19, 2024 14:47 Inactive

@camertron camertron left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice! 🎉

@owenniblock owenniblock left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! A couple of thoughts / questions but non-blocking.

Comment thread generated/components.json Outdated
Comment thread packages/react/package.json Outdated
Comment thread packages/react/src/drafts/MarkdownToolbar/MarkdownToolbar.tsx Outdated
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 20, 2024 10:57 Inactive
@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 20, 2024 11:27 Inactive
@keithamus
keithamus enabled auto-merge March 20, 2024 12:26
@maximedegreve

Copy link
Copy Markdown
Contributor

This is a great addition! There are a few things I've spotted while reviewing this work in Storybook.

  • While not directly part of this component I've noticed that in Safari the <TextArea> isn't styled.
  • The padding seems to be missing in the parent of the buttons.
  • There should be no spacing between the buttons
Showing before and after

@github-actions
github-actions Bot temporarily deployed to storybook-preview-4345 March 20, 2024 12:55 Inactive
@keithamus keithamus added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 21, 2024

@mperrotti mperrotti left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

The issue this resolves (https://github.com/github/primer/issues/3107) mentions a "CommentEditor". What is that?

I'm wondering if we really want this in PRC since we've deprecated similar components:

  • MarkdownEditor
  • MarkdownViewer

@keithamus

Copy link
Copy Markdown
Contributor Author

The issue this resolves (github/primer#3107) mentions a "CommentEditor". What is that?

That is the comment editing experience in dotcom. I'm happy to not include this in PRC and just keep it to dotcom.

@keithamus

Copy link
Copy Markdown
Contributor Author

We'll keep this in the monolith.

@keithamus keithamus closed this Apr 23, 2024
auto-merge was automatically disabled April 23, 2024 15:35

Pull request was closed

@keithamus
keithamus deleted the markdown-toolbar-draft branch April 23, 2024 15:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

update snapshots 🤖 Command that updates VRT snapshots on the pull request

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants