Skip to content

Adding aria-invalid="true" to TextInput when error - #4349

Merged
jonrohan merged 2 commits into
mainfrom
add_aria_invalid_to_textinput
Mar 4, 2024
Merged

Adding aria-invalid="true" to TextInput when error#4349
jonrohan merged 2 commits into
mainfrom
add_aria_invalid_to_textinput

Conversation

@jonrohan

@jonrohanjonrohan commented Mar 4, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3076

Changelog

New

Changed

When validationStatus is of type error add the aria-invalid attribute to the rendered text input.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Mar 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: da0f747

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jonrohan
jonrohan marked this pull request as ready for review March 4, 2024 18:06
@jonrohan
jonrohan requested review from a team and pksjceMarch 4, 2024 18:06
@github-actions

github-actionsBot commented Mar 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.58 KB (+0.02% 🔺)
packages/react/dist/browser.umd.js114.25 KB (+0.03% 🔺)

@pksjcepksjce left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM 👍

@jonrohan
jonrohan added this pull request to the merge queueMar 4, 2024
Merged via the queue into main with commit 15c078dMar 4, 2024
@jonrohan
jonrohan deleted the add_aria_invalid_to_textinput branch March 4, 2024 18:33
@primerprimerBot mentioned this pull request Mar 4, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* Adding aria-invalid="true" to TextInput when error
* Create gold-baboons-live.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@pksjce