Skip to content

Remove redundant aria-label attribute from TextInput.Action - #4354

Merged
broccolinisoup merged 3 commits into
mainfrom
fix-textinput-action-redundant-label
Mar 5, 2024
Merged

Remove redundant aria-label attribute from TextInput.Action#4354
broccolinisoup merged 3 commits into
mainfrom
fix-textinput-action-redundant-label

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

@changeset-bot

changeset-botBot commented Mar 5, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 219d9ed

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 5, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.57 KB (-0.02% 🔽)
packages/react/dist/browser.umd.js114.25 KB (-0.01% 🔽)

@siddharthkpsiddharthkp self-assigned this Mar 5, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review March 5, 2024 15:57
@siddharthkp
siddharthkp requested a review from a teamMarch 5, 2024 15:57
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4354 March 5, 2024 16:01 Inactive
@broccolinisoup
broccolinisoup added this pull request to the merge queueMar 5, 2024
Merged via the queue into main with commit 4b823ffMar 5, 2024
@broccolinisoup
broccolinisoup deleted the fix-textinput-action-redundant-label branch March 5, 2024 22:45
@primerprimerBot mentioned this pull request Mar 5, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* remove aria-label attribute from TextInput.Action
* update snapshot
* Create mighty-pears-deny.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@siddharthkp@broccolinisoup@joshblack