Skip to content

Bug: aria-invalid is overriden in TextInput - #4365

Merged
pksjce merged 5 commits into
mainfrom
fix-text-input-aria-invalid
Mar 7, 2024
Merged

Bug: aria-invalid is overriden in TextInput#4365
pksjce merged 5 commits into
mainfrom
fix-text-input-aria-invalid

Conversation

@pksjce

Copy link
Copy Markdown
Contributor

For release 36.11.0, we see that integration tests are failing.
If we look closer into https://github.com/github/github/actions/runs/8180619423/job/22369116712?pr=315110
the root cause seems to be that the provided aria-invalid prop is being overriden by TextInput. This is introduced in 15c078d#diff-02ef2a0d53f758cb1d0e24c286118bae352cfdae3d88635258903638aca68d04

This PR will change the order of the props to fix it. No visual changes.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@pksjce
pksjce requested a review from a team as a code ownerMarch 7, 2024 06:04
@pksjce
pksjce requested a review from langermankMarch 7, 2024 06:04
@changeset-bot

changeset-botBot commented Mar 7, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 27dacd3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pksjce
pksjce requested a review from jonrohanMarch 7, 2024 06:08
@github-actions

github-actionsBot commented Mar 7, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.71 KB (-0.01% 🔽)
packages/react/dist/browser.umd.js114.38 KB (-0.01% 🔽)

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Good catch!

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Ah, the inputProps splat overriding 👍🏻 ⚡ thanks for the fix

@pksjce
pksjce added this pull request to the merge queueMar 7, 2024
Merged via the queue into main with commit fe9f8baMar 7, 2024
@pksjce
pksjce deleted the fix-text-input-aria-invalid branch March 7, 2024 16:34
@primerprimerBot mentioned this pull request Mar 7, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* Add failing test
* Add failing test
* Add fix
* Create grumpy-shirts-march.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@pksjce@jonrohan@siddharthkp