Skip to content

Bug fix: make active label bold in ActionList - #4392

Merged
langermank merged 5 commits into
mainfrom
fix-navlist-active-bold
Mar 22, 2024
Merged

Bug fix: make active label bold in ActionList#4392
langermank merged 5 commits into
mainfrom
fix-navlist-active-bold

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Closes#4367

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@langermank
langermank requested a review from a team as a code ownerMarch 14, 2024 16:05
@changeset-bot

changeset-botBot commented Mar 14, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d72c78d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 14, 2024
@github-actions

github-actionsBot commented Mar 14, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js88.59 KB (+0.02% 🔺)
packages/react/dist/browser.umd.js88.84 KB (-0.17% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4392 March 14, 2024 16:09 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 14, 2024

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks like we don't have any visual snapshot tests for NavList but the storybook preview looks good! Ship it :shipit:

@siddharthkpsiddharthkp added skip changeset This change does not need a changelog and removed skip changeset This change does not need a changelog labels Mar 14, 2024
@siddharthkp

Copy link
Copy Markdown
Member

@langermank Hi! I'll leave the decision of adding a changelog or not to you.

FYI, merging to main is temporarily disabled

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4392 March 22, 2024 21:58 Inactive
@langermank
langermank added this pull request to the merge queueMar 22, 2024
Merged via the queue into main with commit f100683Mar 22, 2024
@langermank
langermank deleted the fix-navlist-active-bold branch March 22, 2024 22:13
@primerprimerBot mentioned this pull request Mar 22, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* make bold
* Create wicked-rings-lay.md
* snaps
---------
Co-authored-by: Josh Black <joshblack@github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

NavList.Item active state should show up as bold, but doesn't.

4 participants

@langermank@siddharthkp@camertron@joshblack