Skip to content

feat(SelectPanel): add support for announcements to SelectPanel.Loading - #4407

Merged
joshblack merged 5 commits into
mainfrom
feat/add-announce-to-select-panel-loading
Mar 19, 2024
Merged

feat(SelectPanel): add support for announcements to SelectPanel.Loading#4407
joshblack merged 5 commits into
mainfrom
feat/add-announce-to-select-panel-loading

Conversation

@joshblack

Copy link
Copy Markdown
Member

Context: https://github.com/github/primer/issues/3104

Add support for live region announcements to SelectPanel.Loading

Changelog

New

  • Add test file for SelectPanel.Loading

Changed

  • Update type signature for SelectPanel.Loading to have optional children
  • Add support for announcements using Status helper

Removed

Rollout strategy

  • Minor release

Testing & Reviewing

  • Verify that stories that use SelectPanel.Loading announce the given (or default) loading message

@joshblack
joshblack requested a review from a team as a code ownerMarch 18, 2024 17:33
@changeset-bot

changeset-botBot commented Mar 18, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6f0558a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 18, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js87.42 KB (0%)
packages/react/dist/browser.umd.js87.75 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4407 March 18, 2024 17:36 Inactive

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Love it!

Left a tiny suggestion for changelog

Comment thread.changeset/rare-moose-destroy.md Outdated
const SelectPanelLoading= ({children = 'Fetching items...'}: React.PropsWithChildren) => {
return (
<Box
<Status

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Low how simply this looks!

joshblackand others added 2 commits March 19, 2024 10:21
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
@joshblack
joshblack enabled auto-merge March 19, 2024 16:34
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4407 March 19, 2024 16:37 Inactive
@joshblack
joshblack added this pull request to the merge queueMar 19, 2024
Merged via the queue into main with commit 4a7e44aMar 19, 2024
@joshblack
joshblack deleted the feat/add-announce-to-select-panel-loading branch March 19, 2024 16:55
@primerprimerBot mentioned this pull request Mar 19, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
…ng (#4407)
* feat(SelectPanel): add support for announcements to SelectPanel.Loading
* chore: add changeset
* chore: run format
* Update .changeset/rare-moose-destroy.md
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@siddharthkp