Skip to content

chore(deps): update octicons-react and add range for primitives - #4417

Merged
joshblack merged 5 commits into
mainfrom
feat/update-primer-deps
Mar 20, 2024
Merged

chore(deps): update octicons-react and add range for primitives#4417
joshblack merged 5 commits into
mainfrom
feat/update-primer-deps

Conversation

@joshblack

Copy link
Copy Markdown
Member

Update @primer/octicons-react to the latest version. This PR also adds a range to both @primer/octicons-react and @primer/primitives to improve support downstream with monorepos. On our end, in package.json files we should specify the minimum version required for the package so that downstream users can more easily dedupe package versions.

Changelog

New

Changed

  • Update @primer/octicons-react to 19.9.0
  • Add ^ range operator to @primer dependencies

Removed

Rollout strategy

  • Minor release

Testing & Reviewing

  • Verify that icon changes don't introduce a regression in some way
  • Verify that snapshots don't change meaningfully

@joshblack
joshblack requested a review from a team as a code ownerMarch 19, 2024 20:00
@changeset-bot

changeset-botBot commented Mar 19, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ed68403

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Yay thank you! This bit me today in dotcom 😅

@github-actions

github-actionsBot commented Mar 19, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js87.56 KB (+0.03% 🔺)
packages/react/dist/browser.umd.js87.94 KB (+0.15% 🔺)

@camertroncamertron left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Awesome, thank you ❤️

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4417 March 20, 2024 16:56 Inactive
@joshblack
joshblack added this pull request to the merge queueMar 20, 2024
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Mar 20, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4417 March 20, 2024 17:36 Inactive
@joshblack
joshblack added this pull request to the merge queueMar 20, 2024
Merged via the queue into main with commit d7f4f7cMar 20, 2024
@joshblack
joshblack deleted the feat/update-primer-deps branch March 20, 2024 17:56
@primerprimerBot mentioned this pull request Mar 20, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* chore(deps): update octicons-react and add range for primitives
* chore: add changeset
* chore(project): update lockfile
* chore(test): update snapshots
* chore: try to increase timeout limits for ci
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants

@joshblack@pksjce@camertron@broccolinisoup@langermank