Skip to content

ActionBar: Add stories for couple of edge cases - #4424

Merged
pksjce merged 6 commits into
mainfrom
pk/action-bar-edgecases
Apr 2, 2024
Merged

ActionBar: Add stories for couple of edge cases#4424
pksjce merged 6 commits into
mainfrom
pk/action-bar-edgecases

Conversation

@pksjce

Copy link
Copy Markdown
Contributor

Closes #https://github.com/github/primer/issues/3133 and #https://github.com/github/primer/issues/3134

Changelog

Adding two stories for issues that were a problem with the webcomponent in the rails version of this component.

@pksjcepksjce added the skip changeset This change does not need a changelog label Mar 22, 2024
@pksjce
pksjce requested a review from a team as a code ownerMarch 22, 2024 01:19
@pksjce
pksjce requested a review from camertronMarch 22, 2024 01:19
@changeset-bot

changeset-botBot commented Mar 22, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 59b943a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pksjce
pksjce requested a review from siddharthkpMarch 22, 2024 01:19
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4424 March 22, 2024 01:23 Inactive
@github-actions

github-actionsBot commented Mar 22, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js88.62 KB (0%)
packages/react/dist/browser.umd.js88.82 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4424 March 22, 2024 11:21 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4424 March 22, 2024 11:27 Inactive

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Noticing that this PR in addition to adding stories, also moves the overflow to an ActionMenu.

Is that intentional? If yes, do we need a changeset for that?

@pksjce

Copy link
Copy Markdown
ContributorAuthor

Thanks for the review! I should not have made the assumption that draft components don't need changesets.
But we can make larger refactors and API change in this component without needing a major right? Just a patch changeset to document the change to the component should be enough?

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4424 March 24, 2024 23:27 Inactive
@pksjcepksjce removed the skip changeset This change does not need a changelog label Mar 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4424 March 25, 2024 03:17 Inactive
@siddharthkp

Copy link
Copy Markdown
Member

I should not have made the assumption that draft components don't need changesets.

I see! If it's not used anywhere yet, then I'd say it's nice to have a changeset but optional for now.

But we can make larger refactors and API change in this component without needing a major right? Just a patch changeset to document the change to the component should be enough?

patch/minor sounds good in this case :)

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Ship it!

@pksjce
pksjce added this pull request to the merge queueApr 2, 2024
Merged via the queue into main with commit 392a3caApr 2, 2024
@pksjce
pksjce deleted the pk/action-bar-edgecases branch April 2, 2024 00:36
@primerprimerBot mentioned this pull request Apr 2, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@pksjce@siddharthkp