Skip to content

fix(DataTable): add position: relative to scrollable region - #4440

Merged
joshblack merged 3 commits into
mainfrom
fix/add-position-relative-to-scrollable-region
Mar 26, 2024
Merged

fix(DataTable): add position: relative to scrollable region#4440
joshblack merged 3 commits into
mainfrom
fix/add-position-relative-to-scrollable-region

Conversation

@joshblack

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/2865

Update our internal ScrollableRegion component to include position: relative to prevent blowout when an element with position: absolute is included (like with sr-only classes)

Changelog

New

Changed

  • Add position: relative to ScrollableRegion

Removed

Rollout strategy

  • Patch release

Testing & Reviewing

  • Visit the With Row Action Menu story in Storybook
  • Verify that resizing the window that overflow only occurs on the table and not the page

@joshblack
joshblack requested a review from jonrohanMarch 25, 2024 17:46
@joshblack
joshblack requested a review from a team as a code ownerMarch 25, 2024 17:46
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4d98876

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js88.59 KB (-0.04% 🔽)
packages/react/dist/browser.umd.js88.86 KB (+0.04% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4440 March 25, 2024 17:51 Inactive
@joshblack
joshblack added this pull request to the merge queueMar 26, 2024
Merged via the queue into main with commit 674437dMar 26, 2024
@joshblack
joshblack deleted the fix/add-position-relative-to-scrollable-region branch March 26, 2024 14:56
@primerprimerBot mentioned this pull request Mar 26, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* fix(DataTable): add position: relative to scrollable region
* chore: add changeset
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@camertron