Skip to content

ActionBar: Move to main directory (experimental → alpha) - #4574

Merged
siddharthkp merged 9 commits into
mainfrom
actionbar-alpha
May 21, 2024
Merged

ActionBar: Move to main directory (experimental → alpha)#4574
siddharthkp merged 9 commits into
mainfrom
actionbar-alpha

Conversation

@pksjce

Copy link
Copy Markdown
Contributor

Closes #https://github.com/github/primer/issues/3020

Changelog

Make the component be imported from @primer/react and drafts. In next major release, we need to remove it from drafts

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@pksjce
pksjce requested a review from a team as a code ownerMay 9, 2024 02:45
@pksjce
pksjce requested a review from jonrohanMay 9, 2024 02:45
@changeset-bot

changeset-botBot commented May 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f183b91

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented May 10, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js89.29 KB (+0.87% 🔺)
packages/react/dist/browser.umd.js89.59 KB (+0.85% 🔺)

Fix imports
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4574 May 10, 2024 11:52 Inactive

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🎉

export * from '../TooltipV2'
export {default as ActionBar} from './ActionBar'
export * from './ActionBar'
export * from '../ActionBar'

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Kudos for backward compatibility for imports!

"status": "draft",
"a11yReviewed": false,
"status": "alpha",
"a11yReviewed": true,

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nice!

@siddharthkpsiddharthkp changed the title ActionBar: Move to main directoryActionBar: Move to main directory (experimental → alpha)May 21, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4574 May 21, 2024 08:40 Inactive
@siddharthkp
siddharthkp enabled auto-merge May 21, 2024 09:29
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4574 May 21, 2024 09:32 Inactive
@siddharthkp
siddharthkp added this pull request to the merge queueMay 21, 2024
Merged via the queue into main with commit 9c01a93May 21, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@pksjce@jonrohan@siddharthkp