Skip to content

fix(react): update minimum version for @primer/live-region-element - #4575

Merged
joshblack merged 5 commits into
mainfrom
joshblack-patch-1
May 9, 2024
Merged

fix(react): update minimum version for @primer/live-region-element#4575
joshblack merged 5 commits into
mainfrom
joshblack-patch-1

Conversation

@joshblack

Copy link
Copy Markdown
Member

Update our dependency on @primer/live-region-element to the version that addresses ESM module interop errors when using older version of the package.

Changelog

New

Changed

  • Update range for @primer/live-region-element to ^0.6.1

Removed

Rollout strategy

  • Patch release

@joshblack
joshblack requested a review from a team as a code ownerMay 9, 2024 18:38
@joshblack
joshblack requested a review from TylerJDevMay 9, 2024 18:38
@changeset-bot

changeset-botBot commented May 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 71358ff

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented May 9, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js88.37 KB (0%)
packages/react/dist/browser.umd.js88.68 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4575 May 9, 2024 18:45 Inactive
@joshblack
joshblack enabled auto-merge May 9, 2024 18:50
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4575 May 9, 2024 22:07 Inactive
@joshblack
joshblack added this pull request to the merge queueMay 9, 2024
Merged via the queue into main with commit 4ecbe1aMay 9, 2024
@joshblack
joshblack deleted the joshblack-patch-1 branch May 9, 2024 22:24
@primerprimerBot mentioned this pull request May 9, 2024
JelloBagel pushed a commit that referenced this pull request May 16, 2024
…4575)
* fix(react): update minimum version for @primer/live-region-element
* chore(deps): update lockfile and add changeset
* chore: change package versions to use range
* test: update test to run timers
* chore: revert order change when debugging
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@keithamus