Skip to content

fix(Stack): rename spread to justify - #4591

Merged
joshblack merged 3 commits into
mainfrom
4577-justify-no-effect
May 16, 2024
Merged

fix(Stack): rename spread to justify#4591
joshblack merged 3 commits into
mainfrom
4577-justify-no-effect

Conversation

@joshblack

Copy link
Copy Markdown
Member

Closes#4577

This seemed to be lost in the transition, but at one point we renamed spread to justify. This PR updates the CSS from using spread to justify as a result and also updates all snapshots that use spread in the name to justify

Changelog

New

Changed

  • Change CSS that referred to data-spread to data-justify
  • Update tests to call out to justify instead of spread, regenerate snapshots

Removed

Rollout strategy

  • Patch release

Testing & Reviewing

  • Verify that the justify prop correctly updates the component in the storybook playground for Stack

@joshblack
joshblack requested review from a team as code ownersMay 14, 2024 15:39
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a43e3e4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblack
joshblack requested a review from langermankMay 14, 2024 15:39
@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js88.34 KB (0%)
packages/react/dist/browser.umd.js88.65 KB (0%)

@joshblack
joshblack added this pull request to the merge queueMay 16, 2024
Merged via the queue into main with commit 6cce7b7May 16, 2024
@joshblack
joshblack deleted the 4577-justify-no-effect branch May 16, 2024 16:53
@primerprimerBot mentioned this pull request May 16, 2024
JelloBagel pushed a commit that referenced this pull request May 16, 2024
* fix(Stack): rename spread to justify
* test(e2e): regenerate snapshots with spread changed to justify
* chore: add changeset
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Stackjustify prop has no effect

2 participants

@joshblack@langermank