Skip to content

Ensure tab-container uses latest, semver range - #4596

Merged
keithamus merged 3 commits into
mainfrom
ensure-tab-container-uses-latest-semver-range
May 17, 2024
Merged

Ensure tab-container uses latest, semver range#4596
keithamus merged 3 commits into
mainfrom
ensure-tab-container-uses-latest-semver-range

Conversation

@keithamus

Copy link
Copy Markdown
Contributor

Closes #

Changelog

New

Changed

Update to latest tab container element.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@keithamus
keithamus requested a review from a team as a code ownerMay 15, 2024 13:31
@keithamus
keithamus requested a review from pksjceMay 15, 2024 13:31
@changeset-bot

changeset-botBot commented May 15, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d075090

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented May 15, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js88.49 KB (0%)
packages/react/dist/browser.umd.js88.84 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4596 May 17, 2024 09:34 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4596 May 17, 2024 10:37 Inactive
@keithamus
keithamus added this pull request to the merge queueMay 17, 2024
Merged via the queue into main with commit 4fa4faeMay 17, 2024
@keithamus
keithamus deleted the ensure-tab-container-uses-latest-semver-range branch May 17, 2024 11:19
@primerprimerBot mentioned this pull request May 17, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@keithamus@joshblack