Skip to content

Update Banner.tsx - #4598

Merged
lukasoppermann merged 3 commits into
mainfrom
lukasoppermann-rm-space-token-use
May 23, 2024
Merged

Update Banner.tsx#4598
lukasoppermann merged 3 commits into
mainfrom
lukasoppermann-rm-space-token-use

Conversation

@lukasoppermann

@lukasoppermannlukasoppermann commented May 16, 2024

Copy link
Copy Markdown
Contributor

Remove the use of --space tokens

Related to primer/primitives#956

Changelog

Replaced space token with equal base size token.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

It should have no impact on the users.

Testing & Reviewing

Merge checklist

Remove the use of --space tokens
@lukasoppermann
lukasoppermann requested a review from a team as a code ownerMay 16, 2024 08:12
@changeset-bot

changeset-botBot commented May 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a39b256

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented May 16, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js89.35 KB (0%)
packages/react/dist/browser.umd.js89.67 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4598 May 16, 2024 08:16 Inactive
@lukasoppermannlukasoppermann added the skip changeset This change does not need a changelog label May 16, 2024
@joshblack
joshblack added this pull request to the merge queueMay 20, 2024
@joshblack
joshblack removed this pull request from the merge queue due to a manual request May 20, 2024
@joshblack

Copy link
Copy Markdown
Member

@lukasoppermann when you have a sec, could you add a changeset as a patch that talks about the changes to the CSS Custom Properties used? Thanks!

@joshblackjoshblack removed the skip changeset This change does not need a changelog label May 20, 2024
@lukasoppermann

Copy link
Copy Markdown
ContributorAuthor

@joshblack do we want a release for this? There is no change to the output.

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@lukasoppermann@joshblack