Skip to content

ActionMenu: Make sure event handlers on trigger are called - #4648

Merged
siddharthkp merged 8 commits into
mainfrom
fix/actionmenu-button-onclick
Sep 13, 2024
Merged

ActionMenu: Make sure event handlers on trigger are called#4648
siddharthkp merged 8 commits into
mainfrom
fix/actionmenu-button-onclick

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Jun 6, 2024

Copy link
Copy Markdown
Member

@changeset-bot

changeset-botBot commented Jun 6, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3e96d91

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jun 6, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.14 KB (+0.07% 🔺)
packages/react/dist/browser.umd.js97.59 KB (+0.14% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4648 June 6, 2024 16:40 Inactive
@siddharthkpsiddharthkp self-assigned this Jun 18, 2024
@siddharthkpsiddharthkp added patch release bug fixes, docs, housekeeping integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Jun 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4648 June 18, 2024 11:04 Inactive
@siddharthkpsiddharthkp changed the title wip: ActionMenu: Make sure event handlers on trigger are calledActionMenu: Make sure event handlers on trigger are calledJun 18, 2024
@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@github-actions
github-actionsBot deleted the fix/actionmenu-button-onclick branch August 26, 2024 10:05
@siddharthkp
siddharthkp restored the fix/actionmenu-button-onclick branch September 6, 2024 12:49
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

@siddharthkpsiddharthkp reopened this Sep 6, 2024
@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4648 September 6, 2024 12:53 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/341579

@siddharthkpsiddharthkp added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 11, 2024
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

No breaking changes confirmed!

@siddharthkp
siddharthkp marked this pull request as ready for review September 12, 2024 09:47
@siddharthkp
siddharthkp requested a review from a team as a code ownerSeptember 12, 2024 09:47
@siddharthkp
siddharthkp added this pull request to the merge queueSep 13, 2024
Merged via the queue into main with commit c6931d2Sep 13, 2024
@siddharthkp
siddharthkp deleted the fix/actionmenu-button-onclick branch September 13, 2024 10:02
@primerprimerBot mentioned this pull request Sep 13, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
* wip: add mergeAnchorHandlers
* merged onClick and onKeyDown in ActionMenu.Anchor
* improve types
* Create green-schools-smell.md
* cover cases with Tooltip
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghpatch releasebug fixes, docs, housekeepingstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

[ActionMenu.Button] onClick doesn't function

2 participants

@siddharthkp@joshblack