Skip to content

fix: hover background in ActionList.Item - #4693

Merged
langermank merged 3 commits into
primer:mainfrom
renbaoshuo:patch-1
Nov 5, 2024
Merged

fix: hover background in ActionList.Item#4693
langermank merged 3 commits into
primer:mainfrom
renbaoshuo:patch-1

Conversation

@renbaoshuo

Copy link
Copy Markdown
Contributor

When using styled-components 6.x, the background when hovering is not correct.

Changelog

New

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

When using `styled-components` 6.x, the background when hovering is not correct.
@renbaoshuo
renbaoshuo requested a review from a team as a code ownerJune 21, 2024 01:55
@changeset-bot

changeset-botBot commented Jun 21, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7ebfeca

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@renbaoshuo

Copy link
Copy Markdown
ContributorAuthor

ping

@langermank
langermank added this pull request to the merge queueNov 5, 2024
Merged via the queue into primer:main with commit 570d4b3Nov 5, 2024
@primerprimerBot mentioned this pull request Nov 5, 2024
@renbaoshuo
renbaoshuo deleted the patch-1 branch February 5, 2025 12:43
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@renbaoshuo@mperrotti@langermank